)]}'
{"id":"openstack%2Freleases~709891","triplet_id":"openstack%2Freleases~master~I9520c1504bb944718070ef4124a95f38352920e6","project":"openstack/releases","branch":"master","topic":"rocky-em","hashtags":[],"change_id":"I9520c1504bb944718070ef4124a95f38352920e6","subject":"[ironic] Transition Rocky to EM","status":"MERGED","created":"2020-02-26 01:00:20.000000000","updated":"2020-03-09 13:44:47.000000000","submitted":"2020-03-09 13:44:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709891-1583761487983-30632db2","meta_rev_id":"a1c4ababfb54be65e4bd6bf61a4900394c31b741","_number":709891,"virtual_id_number":709891,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 13:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-09 13:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-03-09 09:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-09 13:31:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 01:35:11.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-02-26 18:39:52.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-02-27 10:54:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-02-29 14:48:15.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-02 12:54:19.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-03-09 09:57:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-09 13:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f55d84cb00210545f84ae26375c5090d33c76d21","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 01:00:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32cdaad2ae579773825d1de1736de063039cdced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:17:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81651f71249b4debae4aef75e77b893e : SUCCESS in 12m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6edd6a664c01433eaf39c18825256110 : SUCCESS in 6m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6d23fa2c4b6247b5983a3048af3eccf6 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1d573feb1ac60b893629e90635df907a69c21b65","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-26 17:08:21.000000000","message":"Patch Set 1: Code-Review-1\n\nMaybe networking-generic-switch, python-ironic-inspector-client, and sushi should be included?","accounts_in_message":[],"_revision_number":1},{"id":"7af3aeb440272e09e72143702a7bb3929cb7618a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 17:24:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f57f9fb846ce9cd977e14bbef58a94da5a132b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 17:38:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef04eaed3574afb819082548dc93948 : SUCCESS in 13m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d396633e7ff645cd9346de1a799e5ef1 : SUCCESS in 9m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cb9ea934ad854773bc569a78d3af83f2 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":2},{"id":"df08eb5743751e3d2892864adceb996d0274da7b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-26 18:15:45.000000000","message":"Patch Set 2: Code-Review-1\n\nShould we also remove, like we did with other patches, the file networking-generic-switch-tempest-plugin ?","accounts_in_message":[],"_revision_number":2},{"id":"d5e640d055f324eefd200d5ef56a1da255467626","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-26 18:38:56.000000000","message":"Removed Code-Review-1 by Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"c94a43bd77470530473811d7c9138faa77f1bf72","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-26 18:39:52.000000000","message":"Patch Set 2: Code-Review+1\n\nlist is good. Waiting for ptl or release liaison to check if there is something to backport, for me to update my vote.","accounts_in_message":[],"_revision_number":2},{"id":"d5556cacd22174801adc1460abf6114e78b1e798","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-02-27 10:54:09.000000000","message":"Patch Set 2:\n\nFor unreleased but merged patches, see:\n\nhttps://etherpad.openstack.org/p/ironic-stable-rocky-em","accounts_in_message":[],"_revision_number":2},{"id":"727b8260a80f215acee955654da92f4951b452d9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 14:29:01.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM in the current state, waiting for updates/validations.","accounts_in_message":[],"_revision_number":2},{"id":"ad52cd9a590d8a822f67ed7f0dc0f556e59f149e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-28 08:30:29.000000000","message":"Removed Code-Review+1 by Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"dcc08c3d030e4bf9354a1f205344d774d2f0ccfd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-29 05:07:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e38a756abf4421c032f4b499acb22c889d272ce7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-29 14:29:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eaf30c03b770d31389bd78e45a2c280cb95939b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 14:42:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686807db87e94d688412c3fb68a23337 : SUCCESS in 12m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8cb7d43f7534a54840891543a376b4c : SUCCESS in 7m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e85d3f56b23f45b1a09ffaa8832e0241 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":3},{"id":"a4af888e1a4881de53293763136be626ff6f57b5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-29 14:48:15.000000000","message":"Patch Set 3: Code-Review+1\n\nlgtm from tempest plugin perspective.","accounts_in_message":[],"_revision_number":3},{"id":"fd1e6c97c9505ad29749aa2bdd3e0d764ef8ba5d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-02 10:03:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f5f1d7ff9d520fa9e8ff519808240edb71256bd2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-02 12:54:19.000000000","message":"Patch Set 3: Code-Review-1\n\nIronic seems to have a lot of unreleased changes, lemme check again.","accounts_in_message":[],"_revision_number":3},{"id":"61ee2093ccf6c9bfa022597e440841eb8fc27dc3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-04 18:23:07.000000000","message":"Patch Set 3:\n\n\u003e Ironic seems to have a lot of unreleased changes, lemme check\n \u003e again.\n\nAny updates? We should get this wrapped up soon.","accounts_in_message":[],"_revision_number":3},{"id":"2835c5853e4f530e8bc9061484a984f4e360fe89","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-04 18:50:25.000000000","message":"Patch Set 3:\n\nSure. We need to recheck https://review.opendev.org/706786 into merging, then we can make the final release of ironic and proceed with the EM. Hopefully we can do it all tomorrow.","accounts_in_message":[],"_revision_number":3},{"id":"e9d0789a123be723602f5a840e48a706722c084a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-05 16:28:19.000000000","message":"Patch Set 3:\n\nThe release patch is https://review.opendev.org/#/c/706268/","accounts_in_message":[],"_revision_number":3},{"id":"d2ae8de39b025d2afb53def225e8ed25a39e8625","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-03-09 09:53:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"086e3316c11683811fc8f245ab558e83a26ce37c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-09 09:57:30.000000000","message":"Patch Set 4: Code-Review+2\n\nProperly synced with https://review.opendev.org/#/c/706268 and rebased LGTM.","accounts_in_message":[],"_revision_number":4},{"id":"9ed6b463685d2f5a661cea1cceca6929649eac6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 10:12:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa7322eb1aa4cb4a80e3a09473ff342 : SUCCESS in 19m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4274f0ae93945b68ed430e67bd7ee4d : SUCCESS in 7m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2589781a80e0411aa92adf56fb239986 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":4},{"id":"7e0643d87ad808242b288e8cfe49c3863d128c41","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-09 13:31:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"f4432e5667b2638eb12ea3cdffe4d158c6d886b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:31:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2beaf04df188b348c44e3e0fab1324b35217d466","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:44:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2d4732f934252f18b9d9bbed4245124041750ae8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:44:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/638cb1de1e6045a0880691d9359da281 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fbd60cd64348cc9893bc965e683638 : SUCCESS in 11m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8bb0419eb18947c4bffe593e82d660fa : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2a8708c6db9dd00a81f4f4668cb792329b1446b3","revisions":{"4c4b8d02d68312a81639c227e1462361b842fc43":{"kind":"REWORK","_number":1,"created":"2020-02-26 01:00:20.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/709891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/709891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/709891/1"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"subject":"[ironic] Transition Rocky to EM","message":"[ironic] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I9520c1504bb944718070ef4124a95f38352920e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4b8d02d68312a81639c227e1462361b842fc43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4b8d02d68312a81639c227e1462361b842fc43"}]},"branch":"refs/heads/master"},"c677c2cd03a51a07300b6349e403702d5f9d7e9d":{"kind":"REWORK","_number":2,"created":"2020-02-26 17:24:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/709891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/709891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/709891/2"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 17:23:59.000000000","tz":-360},"subject":"[ironic] Transition Rocky to EM","message":"[ironic] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I9520c1504bb944718070ef4124a95f38352920e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c677c2cd03a51a07300b6349e403702d5f9d7e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c677c2cd03a51a07300b6349e403702d5f9d7e9d"}]},"branch":"refs/heads/master"},"f2a30a616418aa587fec52def06a8c1e14d6f41b":{"kind":"REWORK","_number":3,"created":"2020-02-29 14:29:31.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/709891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/709891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/709891/3"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-29 14:29:27.000000000","tz":-360},"subject":"[ironic] Transition Rocky to EM","message":"[ironic] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I9520c1504bb944718070ef4124a95f38352920e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2a30a616418aa587fec52def06a8c1e14d6f41b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2a30a616418aa587fec52def06a8c1e14d6f41b"}]},"branch":"refs/heads/master"},"2a8708c6db9dd00a81f4f4668cb792329b1446b3":{"kind":"REWORK","_number":4,"created":"2020-03-09 09:53:07.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/91/709891/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/709891/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/709891/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/709891/4"}}},"commit":{"parents":[{"commit":"a47b6b503f8faff81b55732cb7b146670fd2b26f","subject":"Merge \"Release stable branches for ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a47b6b503f8faff81b55732cb7b146670fd2b26f"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-03-09 09:52:47.000000000","tz":60},"subject":"[ironic] Transition Rocky to EM","message":"[ironic] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I9520c1504bb944718070ef4124a95f38352920e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2a8708c6db9dd00a81f4f4668cb792329b1446b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2a8708c6db9dd00a81f4f4668cb792329b1446b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
