)]}'
{"id":"openstack%2Freleases~709896","triplet_id":"openstack%2Freleases~master~Ifeb733b4d137c12176fba556acb00717731a0219","project":"openstack/releases","branch":"master","topic":"rocky-em","hashtags":[],"change_id":"Ifeb733b4d137c12176fba556acb00717731a0219","subject":"[manila] Transition Rocky to EM","status":"MERGED","created":"2020-02-26 01:00:49.000000000","updated":"2020-03-12 16:15:08.000000000","submitted":"2020-03-12 16:15:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709896-1584029708850-7e8b974d","meta_rev_id":"74ef1b8671e958e13d0975c7c33d1db2ff24e702","_number":709896,"virtual_id_number":709896,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-03-12 15:10:28.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-12 16:15:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-12 15:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2020-03-12 16:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-12 15:54:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 01:00:49.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-02-26 01:44:59.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-02-29 14:50:09.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-12 15:10:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-03-12 16:04:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-12 16:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e81d0454e5b3ee10a6350a188c87af60d5f53b3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 01:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"714308cd001333f2b4a711586de6a1c9a283d387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:12:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69642d43e4e84b80ad5bba3bd240bed6 : SUCCESS in 10m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/694e71045b9c4f8da92e1616e5201de0 : SUCCESS in 6m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e1f6e526e949443993faefa04d69183e : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fa939e8e9e12d4042b25eb9e9f5aea60d5cb33b4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 01:18:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bfe63d0a41131a3db6127b2f79cf6d8d56e94e7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-02-26 01:23:15.000000000","message":"Patch Set 2: Code-Review-1\n\nHi Sean, \n\nWe\u0027re trying to get an important fix into manila, and it might be appropriate to hold this for that ... I\u0027ll circle back at the end of the week, post our community meeting\n\n[1] https://wiki.openstack.org/wiki/Manila/Meetings#Next_meeting","accounts_in_message":[],"_revision_number":2},{"id":"8213209a50455a7c1698561f8b3011722cf83a98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:36:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/886bfc3f53a8495c8f8765b25ced6b6e : SUCCESS in 12m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2eedec8d538f441f8008bbad99d36d8e : SUCCESS in 6m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/372b5dadacee45f98f287d3fe2361324 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fb502203a66e05b647eb8232a3cb9034fd665346","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-29 04:44:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nwe need to fix the manila-tempest-plugin release.","accounts_in_message":[],"_revision_number":2},{"id":"cb106065d886155c1f6bd8783d9e8daf005c8f63","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-29 14:32:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2cef3fbb23acb0fe0e8d6d97912fc2e940462c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 14:44:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3bcdc4d665b424bbc57aa8d35f3a27f : SUCCESS in 11m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32ce5af9d1514430b3482d8f21173144 : SUCCESS in 7m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c956491f9c7f4777bb8a79223166b793 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":3},{"id":"bc69bcb1024909df5226775455b5b4f87db89e0e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-29 14:50:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6e6912d362ff10eb2181d9f2109e19723a45490c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-04 12:18:20.000000000","message":"Patch Set 3: Workflow-1\n\nUpdate from team that they are working through an important bugfix that has been held up by issues beyond their control. Holding a little longer on this one.","accounts_in_message":[],"_revision_number":3},{"id":"876d6fb363c0c09a1b23600959c1479ddce74c5c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-04 15:34:48.000000000","message":"Patch Set 3:\n\n\u003e Update from team that they are working through an important bugfix that has been held up by issues beyond their control. Holding a little longer on this one.\n\nThank you Sean!","accounts_in_message":[],"_revision_number":3},{"id":"459f1809db9d6acda4399c67f57aa6cf7edb0701","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-10 20:19:11.000000000","message":"Patch Set 3:\n\n\u003e \u003e Update from team that they are working through an important\n \u003e bugfix that has been held up by issues beyond their control.\n \u003e Holding a little longer on this one.\n \u003e \n \u003e Thank you Sean!\n\nAny updates on this Goutham?","accounts_in_message":[],"_revision_number":3},{"id":"23c6fb9d5d7e65af9467c3e6fd5d60ef9efaf09e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-10 21:24:36.000000000","message":"Patch Set 3:\n\n\u003e Any updates on this Goutham?\n\nHi Sean, thank you for checking. Am waiting on this fix: https://review.opendev.org/#/c/712165/ - will circle back after we merge it. \n\nRelated to CVE-2020-9543\nBug: https://bugs.launchpad.net/manila/+bug/1861485\nOSSA: https://review.opendev.org/#/c/712172/","accounts_in_message":[],"_revision_number":3},{"id":"d8a706c9d0676bc2078c859d0dd5d56b4bdf94b3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-11 16:16:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"74e8a01a2a05c54038e6ffb181fad71dd824fda6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-11 16:18:13.000000000","message":"Patch Set 4: Code-Review+1\n\nThank you for your patience, Sean. This is now ready. \n\n+1 to try and trigger the PTL-Approved flag","accounts_in_message":[],"_revision_number":4},{"id":"804915c1c3014c97ab11a01d0f6dc68959eb073f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-11 16:19:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"474702143df3d5ff80ebadac99b0a4b97808f65e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 16:20:15.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"4ca7d4ace0ebcd2be57f0237ba71b121512cc5a5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-11 16:22:57.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5385d959ab142f81360d47305916017b813a3891","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 16:24:10.000000000","message":"Patch Set 4: -PTL-Approved","accounts_in_message":[],"_revision_number":4},{"id":"d0a2da529f1b0f9c1693e948d16efffa373f52fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 16:29:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ec4b64f14946339583881ace96b676 : SUCCESS in 10m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e53a9a8de254884b689404c281ab789 : FAILURE in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1b9f034c1fc54d19bbb11907e7d90cdb : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":4},{"id":"4f58cabf58562219c9cbc28dac4a768674b71b6d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-11 16:31:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1b93bfc067782cdbcd9c360774aefa06517beec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 16:50:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c4ddb02ba74ebc98b524adbd7016fe : SUCCESS in 10m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de5bd4add44849948360490c31d652f2 : FAILURE in 8m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/095a1e2cddaa41068457a23b2a4cf0c7 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":5},{"id":"7cef7ccf0910c4bcb4fe4761daf7aebb7ec5efbd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-11 16:58:40.000000000","message":"Patch Set 5:\n\nWill just need to recheck once the parent lands and the tag is processed.","accounts_in_message":[],"_revision_number":5},{"id":"9d05edc34c64069253ba1f95e09305b5c9e9cf7c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-12 15:10:28.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3a948d25b5c45e602b1192e91fde1a100f1c2038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 15:25:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009dea2a602045c08e16fe59cb0b78df : SUCCESS in 12m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd1d862655424980965db8895c4854be : SUCCESS in 6m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/71317d787db441688201456455f10cec : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":5},{"id":"0eb8c5b3dd6a266fdd5bbc9e7157a6c1497af276","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-12 15:54:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks Goutham!","accounts_in_message":[],"_revision_number":5},{"id":"1dfda9ff694bdbe3696ccfd8350a926d2eb8a57a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 15:57:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3520bb3f68aeffa086635bf9d909b56db0b90a45","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-12 16:04:05.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f7882a360bbcbafacb82d2865430d55c6fa0bc92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 16:15:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"775c3af5544b68d0d6d41883f7c458fe8f65d3fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 16:15:08.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f47a5b26da2c4e9682f50034518ae3c0 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d535674151e3458fbe28fcb1b5af7781 : SUCCESS in 12m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ecc3b40672b4a4b8fe1758c47b750ab : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"45b9ac3c5e2deee9567279b8344469bf742f001d","revisions":{"3672ec124c8d62b6842577ec292233ff4b0dfe0b":{"kind":"REWORK","_number":1,"created":"2020-02-26 01:00:49.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/96/709896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/709896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/709896/1"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"subject":"[manila] Transition Rocky to EM","message":"[manila] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ifeb733b4d137c12176fba556acb00717731a0219\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3672ec124c8d62b6842577ec292233ff4b0dfe0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3672ec124c8d62b6842577ec292233ff4b0dfe0b"}]},"branch":"refs/heads/master"},"2fff7851a4b8a41cc1d492cd4b498b9a60b4729d":{"kind":"REWORK","_number":2,"created":"2020-02-26 01:18:33.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/96/709896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/709896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/709896/2"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 01:18:29.000000000","tz":-360},"subject":"[manila] Transition Rocky to EM","message":"[manila] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ifeb733b4d137c12176fba556acb00717731a0219\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2fff7851a4b8a41cc1d492cd4b498b9a60b4729d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2fff7851a4b8a41cc1d492cd4b498b9a60b4729d"}]},"branch":"refs/heads/master"},"4362d5fda0c6637d74b52ff09f5f306932482ea1":{"kind":"REWORK","_number":3,"created":"2020-02-29 14:32:33.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/96/709896/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/709896/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/709896/3"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-29 14:32:29.000000000","tz":-360},"subject":"[manila] Transition Rocky to EM","message":"[manila] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ifeb733b4d137c12176fba556acb00717731a0219\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4362d5fda0c6637d74b52ff09f5f306932482ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4362d5fda0c6637d74b52ff09f5f306932482ea1"}]},"branch":"refs/heads/master"},"6993bd8cfcbdce8a6427547b7ef1d51dcc2dedab":{"kind":"REWORK","_number":4,"created":"2020-03-11 16:16:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/96/709896/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/709896/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/709896/4"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-11 16:16:18.000000000","tz":-420},"subject":"[manila] Transition Rocky to EM","message":"[manila] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ifeb733b4d137c12176fba556acb00717731a0219\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6993bd8cfcbdce8a6427547b7ef1d51dcc2dedab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6993bd8cfcbdce8a6427547b7ef1d51dcc2dedab"}]},"branch":"refs/heads/master"},"45b9ac3c5e2deee9567279b8344469bf742f001d":{"kind":"REWORK","_number":5,"created":"2020-03-11 16:31:14.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/96/709896/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/709896/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/709896/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/709896/5"}}},"commit":{"parents":[{"commit":"6782ad2f14cc36f890ebec8c15d79f8b7889e570","subject":"[manila] Release 7.4.1 from stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6782ad2f14cc36f890ebec8c15d79f8b7889e570"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:40:08.000000000","tz":-360},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-11 16:30:58.000000000","tz":-420},"subject":"[manila] Transition Rocky to EM","message":"[manila] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: Ifeb733b4d137c12176fba556acb00717731a0219\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/45b9ac3c5e2deee9567279b8344469bf742f001d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/45b9ac3c5e2deee9567279b8344469bf742f001d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
