)]}'
{"id":"openstack%2Freleases~710220","triplet_id":"openstack%2Freleases~master~Ie5842e4edcdfef4067da3df15cee565196a84564","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ie5842e4edcdfef4067da3df15cee565196a84564","subject":"Release monasca-api 2.7.1","status":"MERGED","created":"2020-02-27 10:40:49.000000000","updated":"2020-03-02 10:51:50.000000000","submitted":"2020-03-02 10:51:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710220-1583146310333-20d59637","meta_rev_id":"97b37ea7c3776c4c9f56fc12a2e4559181e80558","_number":710220,"virtual_id_number":710220,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-02 10:51:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-27 16:13:51.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 17:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-02 10:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-02 10:35:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-27 14:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 17:14:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-02 10:35:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-02 10:51:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3a653b0d0d062566ee1850f9e9250a1b5aba280","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-27 10:40:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a23cfc51c7715f3791835774b3425ef275b36e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 11:13:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac31c72b7c74e9e9fa400b627ee4646 : SUCCESS in 12m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d3431ab3f69b45df917e056edf98c15a : SUCCESS in 6m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b3a5796e32b2446f99d931b479ce7cab : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"63cc60cdad343ebc6f9f431397bf30bcd4d1ee32","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 14:44:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f01bb25618b7d54aefa65159259e1211f1d392cb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:55:06.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"887222aa7487a8c4430d744770b1e38338b01260","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 15:28:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\n-1 to highlight my thinking and discuss about.","accounts_in_message":[],"_revision_number":1},{"id":"d4698c37b584a946b93c981bc7e84b0fa498b7e8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 15:33:50.000000000","message":"Patch Set 1: -Code-Review\n\nCan\u0027t do a minor version bump since we have 2.8.0 in stein.\n\nBut looking closer at those changes, in addition to not being properly backported, changing an API is not allowed for backports, so all of those should be reverted.","accounts_in_message":[],"_revision_number":1},{"id":"b33ae65179ba302a9d5045a456b28c6a35a6a09f","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-27 16:13:51.000000000","message":"Patch Set 1:\n\nI agree that it shouldn\u0027t get backported in the first place. Although we try to follow stable branch policy we technically don\u0027t have the corresponding tag [1]. In that particular case the API change is small and backwards compatible, so I\u0027d like to ask you to make an exception here and accept this release.\n\n[1] https://governance.openstack.org/tc/reference/tags/stable_follows-policy.html","accounts_in_message":[],"_revision_number":1},{"id":"713893401c7bde3d1ada717e05475cc5eca365bf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 17:14:01.000000000","message":"Patch Set 1: Code-Review+2\n\nVery fair point - if this repo doesn\u0027t assert following stable policy, then, while still not recommended, there aren\u0027t as many restrictions against backporting something like this.\n\nAnd if it is still considered compatible with the currently released code, then that actually is really good.\n\nThanks for explaining things Witold.","accounts_in_message":[],"_revision_number":1},{"id":"0dead7461df5c2189fdc55484e27a7b7becab96c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-02 10:35:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d17a57ff1a6b100cee2a73fb61406ea500ba1dec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 10:35:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a79bd87969918aa0ba8e9f10536d210fc0c8556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 10:51:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba8f1a51a08849c3aaa0cf57ea66fc25 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c5bb8498464ec1b346467dd93bab60 : SUCCESS in 13m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9cb41bb776e544d4b0f8e3a477743535 : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":1},{"id":"97b37ea7c3776c4c9f56fc12a2e4559181e80558","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 10:51:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6e6a490222a20fce20355444273219831e1a1ce","revisions":{"c6e6a490222a20fce20355444273219831e1a1ce":{"kind":"REWORK","_number":1,"created":"2020-02-27 10:40:49.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/20/710220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/20/710220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/710220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/710220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/20/710220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/20/710220/1"}}},"commit":{"parents":[{"commit":"cf044a078d3119644bd88e18dbdbba20ed16a181","subject":"Merge \"[zaqar] Transition Rocky to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf044a078d3119644bd88e18dbdbba20ed16a181"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-27 10:40:24.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-27 10:40:24.000000000","tz":60},"subject":"Release monasca-api 2.7.1","message":"Release monasca-api 2.7.1\n\nChange-Id: Ie5842e4edcdfef4067da3df15cee565196a84564\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6e6a490222a20fce20355444273219831e1a1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6e6a490222a20fce20355444273219831e1a1ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
