)]}'
{"id":"openstack%2Freleases~710222","triplet_id":"openstack%2Freleases~master~Id9076e6ae8cd327fbb885fc54f69094f6d45a1a0","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Id9076e6ae8cd327fbb885fc54f69094f6d45a1a0","subject":"Release monasca-ceilometer 1.8.2","status":"ABANDONED","created":"2020-02-27 10:48:42.000000000","updated":"2020-03-02 15:45:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d1183ac95a6449ba4c178e442b35021db799001f","_number":710222,"virtual_id_number":710222,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-27 11:14:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-27 15:44:01.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"date":"2020-02-27 15:23:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":-1,"date":"2020-02-27 15:16:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-02-27 15:37:20.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-27 14:55:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 11:14:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-27 15:16:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-27 15:23:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-27 15:37:20.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"}],"messages":[{"id":"21291c01a58ce19c3421720dee8c1590739c3dc6","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-27 10:48:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e229c6c9082aa4e44440ccef1fda1c5b9bc50f73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 11:14:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab9b8d4848f434fbf15e93ee3131394 : SUCCESS in 13m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ad4433f48984335b96e64a251e685f2 : SUCCESS in 9m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f2f8549d30a24409a0e87ac1dd6d909a : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fb8a9e9bfa775b7451c9adb5489f63a40a77fd82","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 14:44:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"734d0eb53df73baf02e2217cc3cc903732eaf5cc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:55:11.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"c4be122da3d1478abc021aa93bdafb3b1f859388","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 15:16:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nHello,\n\n-1 to highlight my question, I\u0027ll update my mindset after some related discussion.","accounts_in_message":[],"_revision_number":1},{"id":"0a5712dcbf07925931bd8ff0cb23e628871513ad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 15:23:36.000000000","message":"Patch Set 1: -Code-Review\n\nYeah, looking closer at that, that really was not even appropriate to backport. That should be reverted and then this updated with the hash after that.","accounts_in_message":[],"_revision_number":1},{"id":"5abc0a65d54c7bcbddc7921796451dbbf94a5cd4","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-27 15:44:01.000000000","message":"Patch Set 1:\n\nAdded Joseph to reviewers. I think I\u0027m OK with skipping this release. The code is included in 1.9.0 (Stein).","accounts_in_message":[],"_revision_number":1},{"id":"f88ac661962f9ff0ab31827c3ba22a0dc9d390eb","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-03-02 15:02:02.000000000","message":"Abandoned\n\nI think we\u0027re safe to skip this release.","accounts_in_message":[],"_revision_number":1},{"id":"d1183ac95a6449ba4c178e442b35021db799001f","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-03-02 15:45:21.000000000","message":"Patch Set 1:\n\nFor what it\u0027s worth, @hberaud the Gnocchi change you reference was actually removed in the Ceilometer stable/rocky branch, so the removal of the Gnocchi dispatcher_gnocchi options is not a backport in Ceilosca but rather is Ceilosca stable/rocky catching up to the Ceilometer stable/rocky release.  Again, not a backport but trying to get Ceilosca in harmony with Ceilometer in the Rocky release.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e110bbcc0d35c506ab7d9209ee602ad248b1cb78","revisions":{"e110bbcc0d35c506ab7d9209ee602ad248b1cb78":{"kind":"REWORK","_number":1,"created":"2020-02-27 10:48:42.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/22/710222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/710222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/710222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/710222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/710222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/710222/1"}}},"commit":{"parents":[{"commit":"cf044a078d3119644bd88e18dbdbba20ed16a181","subject":"Merge \"[zaqar] Transition Rocky to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf044a078d3119644bd88e18dbdbba20ed16a181"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-27 10:48:29.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-27 10:48:29.000000000","tz":60},"subject":"Release monasca-ceilometer 1.8.2","message":"Release monasca-ceilometer 1.8.2\n\nChange-Id: Id9076e6ae8cd327fbb885fc54f69094f6d45a1a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e110bbcc0d35c506ab7d9209ee602ad248b1cb78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e110bbcc0d35c506ab7d9209ee602ad248b1cb78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
