)]}'
{"id":"openstack%2Freleases~710224","triplet_id":"openstack%2Freleases~master~Ia8193dd3e2d3b758d4e86b5fd503e8086b474469","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ia8193dd3e2d3b758d4e86b5fd503e8086b474469","subject":"Release monasca-persister 1.12.1","status":"MERGED","created":"2020-02-27 10:51:50.000000000","updated":"2020-02-27 16:54:34.000000000","submitted":"2020-02-27 16:54:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710224-1582822474489-28c1a561","meta_rev_id":"80dd23d78f0b0005158c47f1f854c29540241f0a","_number":710224,"virtual_id_number":710224,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 16:54:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 14:45:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-27 16:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-27 16:26:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-27 14:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 14:45:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-27 16:26:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-27 16:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"941bae6bfd4d43f22f02490a9b647152045ae748","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-27 10:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd049a32881c4d3c2476e4899a73c8a85aecb19e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 11:34:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8d14e5b2754a6f9d817f4b12d2f395 : SUCCESS in 14m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b71880145a94210b1e23ed364ddf59b : SUCCESS in 8m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ab4d89ccb8334643826d0312894c9e92 : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":1},{"id":"771f605bdc10d6b0a37824deee8dd7a57c2cb62b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 14:45:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0690552b04c4aca465249e5eeb5e267803be161c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:55:17.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"acc01adc16075eb91ca3074ffd46a11d9a79b91c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 15:08:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc12d697df37dd0f8eb19cdd2562fb155a65115d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 15:20:16.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nI consider that more of a bugfix - one that required the addition of a config option to work around.\n\nWe normally do not want to add new config options in backports. But in this case, the default for the option keeps the behavior consistent with the existing behavior, so it is safe to do.\n\nThe first release in stein was 1.13.0, so we really only have the bugfix version to work with here.","accounts_in_message":[],"_revision_number":1},{"id":"d79ef269c49ac1209f84670683d40249f6162779","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 16:26:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e I consider that more of a bugfix - one that required the addition\n \u003e of a config option to work around.\n \u003e \n \u003e We normally do not want to add new config options in backports. But\n \u003e in this case, the default for the option keeps the behavior\n \u003e consistent with the existing behavior, so it is safe to do.\n \u003e \n \u003e The first release in stein was 1.13.0, so we really only have the\n \u003e bugfix version to work with here.\n\nOk make sense","accounts_in_message":[],"_revision_number":1},{"id":"ac3e468852c19ef73e93d584fbbc29f0cf06aa8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:26:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57496760f978dfcc0de811bbc19abc8f4925fd71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:54:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"80dd23d78f0b0005158c47f1f854c29540241f0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:54:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d670e2ef4e9341dc8abfdef6db28ad05 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d2cd5c092d471f8f67d32fe9c084d4 : SUCCESS in 11m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd17ab1256f94782bad1dc74cf197d17 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bac58b7d70b48b0a292b6385fb88647e2d8268e","revisions":{"4bac58b7d70b48b0a292b6385fb88647e2d8268e":{"kind":"REWORK","_number":1,"created":"2020-02-27 10:51:50.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/24/710224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/710224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/710224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/710224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/710224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/710224/1"}}},"commit":{"parents":[{"commit":"cf044a078d3119644bd88e18dbdbba20ed16a181","subject":"Merge \"[zaqar] Transition Rocky to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf044a078d3119644bd88e18dbdbba20ed16a181"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-27 10:51:39.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-27 10:51:39.000000000","tz":60},"subject":"Release monasca-persister 1.12.1","message":"Release monasca-persister 1.12.1\n\nChange-Id: Ia8193dd3e2d3b758d4e86b5fd503e8086b474469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4bac58b7d70b48b0a292b6385fb88647e2d8268e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4bac58b7d70b48b0a292b6385fb88647e2d8268e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
