)]}'
{"id":"openstack%2Freleases~710878","triplet_id":"openstack%2Freleases~master~If7caf7c0ad6b3599fcf825982c3a0cd6edc3aa0e","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"If7caf7c0ad6b3599fcf825982c3a0cd6edc3aa0e","subject":"Update release liason for Swift","status":"MERGED","created":"2020-03-02 22:08:41.000000000","updated":"2020-03-12 10:18:26.000000000","submitted":"2020-03-12 10:18:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710878-1584008306558-74ea99eb","meta_rev_id":"b6b05a2416c86bbc8ea85bda87cd6cf1e2a5c6d2","_number":710878,"virtual_id_number":710878,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-02 22:23:41.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-03-04 12:02:00.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-12 10:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-12 10:01:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-12 10:01:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 22:23:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-04 12:02:00.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-03-12 10:01:27.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-03-12 10:18:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebd3032da6c24f28880fe993458acb18929e459d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-03-02 22:08:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77783da9264fc763d1edfcab5bd454265c832d53","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-02 22:23:41.000000000","message":"Patch Set 1:\n\nYou could just remove the entry. The PTL is included, so this is just for non-PTL people that have the authority to approve release requests.\n\nThat said, if the PTL role ever changes, you\u0027d be all set to be able to continue filling that role.\n\nMight be good to add someone other than yourself in case you need to delegate for any reason.","accounts_in_message":[],"_revision_number":1},{"id":"1c5c2d7287e0418e7829b9d755820efcddf973cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 22:32:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1d51b7db1724f548d8627f98366a7e7 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92f525554d24f94a28f0a9a762afd9d : SUCCESS in 15m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb17b47f743e45b3bd601e61926b0a7d : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"da1df2dbaf23c2c25b26dc3de1c54d47351b85b4","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-04 12:02:00.000000000","message":"Patch Set 1:\n\nPlease see Sean\u0027s comment, I was planning to do the same comment.","accounts_in_message":[],"_revision_number":1},{"id":"75c536cc3864d773e1a11ff3e4ec38762d138569","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-12 10:01:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nRedundancy does not hurt","accounts_in_message":[],"_revision_number":1},{"id":"4dc3dbe3d42bba6b90cb1cb66535225880ba3e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 10:01:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93a80461cb71f148362e2d47b6238f783dce219f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 10:18:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b6b05a2416c86bbc8ea85bda87cd6cf1e2a5c6d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 10:18:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6c0bd7fa2614322ac0bc5af57af1121 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e688360ca93d457fab4cf3ebf1107b97 : SUCCESS in 11m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc2e91267a8b45eaa72d74442c15bc82 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79b7dceb2471fe03b550e0bc6c24ac9754d06126","revisions":{"79b7dceb2471fe03b550e0bc6c24ac9754d06126":{"kind":"REWORK","_number":1,"created":"2020-03-02 22:08:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/710878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/710878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/710878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/710878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/710878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/710878/1"}}},"commit":{"parents":[{"commit":"a5d71097ea6778e05081ec5263e9bafdad506cd1","subject":"Merge \"Release OpenStack-Ansible Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5d71097ea6778e05081ec5263e9bafdad506cd1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-02 22:07:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-03-02 22:07:50.000000000","tz":-480},"subject":"Update release liason for Swift","message":"Update release liason for Swift\n\nJohn\u0027s gotten tired of needing to forward patches to me ;-)\n\nChange-Id: If7caf7c0ad6b3599fcf825982c3a0cd6edc3aa0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b7dceb2471fe03b550e0bc6c24ac9754d06126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79b7dceb2471fe03b550e0bc6c24ac9754d06126"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
