)]}'
{"id":"openstack%2Freleases~711196","triplet_id":"openstack%2Freleases~master~Ic6c0f4089259cfc79b70b80c571f5b903a04fd7f","project":"openstack/releases","branch":"master","topic":"keystone-liaison","hashtags":[],"change_id":"Ic6c0f4089259cfc79b70b80c571f5b903a04fd7f","subject":"Update keystone liaison\u0027s email","status":"MERGED","created":"2020-03-04 10:29:44.000000000","updated":"2020-03-05 13:58:55.000000000","submitted":"2020-03-05 13:58:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711196-1583416735850-eb552905","meta_rev_id":"9f8069a033f8174a5cb98e6fdf7b41cc89ba1d2a","_number":711196,"virtual_id_number":711196,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 13:58:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"recommended":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2020-03-04 21:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2020-03-05 13:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2020-03-05 13:19:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 21:26:53.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-03-05 13:19:36.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-03-05 13:58:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b48dc6aa789ffed8dc4e2cd03c778aa5582a825a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-04 10:29:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb43da2cfb696c2b5b5ae2886b9d2ffe64b74460","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-04 10:32:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e9350706260dee0db8e45adb80fa4af0ad5df1dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 10:48:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/191aee7da9d645b9b32b1702e88e0422 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cfddf1baaf64a8a886cb5b414b0e2b1 : SUCCESS in 12m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/548dbafc52c645058133d54056da5c90 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"c1e4f602683375c883e8736b669f11626cf03bf3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-04 12:03:41.000000000","message":"Patch Set 2: Code-Review+1\n\nWaiting for Kristi\u0027s approval before +w-ing it.","accounts_in_message":[],"_revision_number":2},{"id":"fd66af7c6132219f7563722de23fd41662afa396","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-03-04 21:26:53.000000000","message":"Patch Set 2: Code-Review+1\n\nOh thanks!","accounts_in_message":[],"_revision_number":2},{"id":"a11273da620e474d2c1576360a18fee4b4e47d16","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-05 13:19:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4855c0bbc841058e911ee6f02bf8fe5e6f439e76","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-05 13:19:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3b8ab0deae25f7ec608a2aec93f566674261db83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 13:19:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d68b852af2644db89871cae76ead52450273347c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 13:58:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f8725b735a4f5387a77510656801a9 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93df11568fda49279a204faf6986bbee : SUCCESS in 11m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f537da1f997441d84799cb01b210778 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":2},{"id":"14bae0b652ae2888a594358562d681b60d7e73ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 13:58:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de22dbefa69ca67f18f1e248011d2cbbab10d79e","revisions":{"d79f836df1b3979c92a8167d38adc7df11f4fc10":{"kind":"REWORK","_number":1,"created":"2020-03-04 10:29:44.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/711196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/711196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/711196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/711196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/711196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/711196/1"}}},"commit":{"parents":[{"commit":"d38dff6844d426306781dc3ec084ad476919f790","subject":"Merge \"Final Cinder team releases for stable/rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d38dff6844d426306781dc3ec084ad476919f790"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-04 10:28:01.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-04 10:28:01.000000000","tz":60},"subject":"Update keystone liaison\u0027s email","message":"Update keystone liaison\u0027s email\n\nKristi Nikolla\u0027s keystone liaison but is email is not recognized.\n\nChange-Id: Ic6c0f4089259cfc79b70b80c571f5b903a04fd7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d79f836df1b3979c92a8167d38adc7df11f4fc10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d79f836df1b3979c92a8167d38adc7df11f4fc10"}]},"branch":"refs/heads/master"},"de22dbefa69ca67f18f1e248011d2cbbab10d79e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-04 10:32:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/96/711196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/96/711196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/711196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/711196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/96/711196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/96/711196/2"}}},"commit":{"parents":[{"commit":"d38dff6844d426306781dc3ec084ad476919f790","subject":"Merge \"Final Cinder team releases for stable/rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d38dff6844d426306781dc3ec084ad476919f790"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-04 10:28:01.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-04 10:32:29.000000000","tz":60},"subject":"Update keystone liaison\u0027s email","message":"Update keystone liaison\u0027s email\n\nKristi Nikolla is the keystone liaison but is email is not recognized.\n\nChange-Id: Ic6c0f4089259cfc79b70b80c571f5b903a04fd7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de22dbefa69ca67f18f1e248011d2cbbab10d79e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/de22dbefa69ca67f18f1e248011d2cbbab10d79e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
