)]}'
{"id":"openstack%2Freleases~711600","triplet_id":"openstack%2Freleases~master~I95d99f9aa1eb850ab8d6d2c1223d92fa8b58dea9","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I95d99f9aa1eb850ab8d6d2c1223d92fa8b58dea9","subject":"Release neutron-lib 2.2.0","status":"MERGED","created":"2020-03-06 09:57:43.000000000","updated":"2020-03-10 14:10:10.000000000","submitted":"2020-03-10 14:10:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711600-1583849410133-72e87165","meta_rev_id":"6ab204ba8c3f3bbeb82994b8da0e262321c08d21","_number":711600,"virtual_id_number":711600,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-03-10 08:29:01.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 14:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-10 13:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-10 10:36:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-03-10 08:40:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-03-10 09:07:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-10 13:48:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-10 08:29:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 10:23:40.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-03-06 11:27:05.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-03-10 08:40:47.000000000","updated_by":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"reviewer":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"state":"REVIEWER"},{"updated":"2020-03-10 09:07:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-10 10:36:45.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-03-10 13:48:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-10 14:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5857e7087fa89ec58b42920188f9610d0fc4f8d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-06 09:57:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60ba3a391cd3a9f2848b1cdd8662da4adefde19f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 10:13:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90d981ec3754997a7919b3a6403f895 : SUCCESS in 11m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a184e30c2aa347b59111ad151b7cb378 : SUCCESS in 5m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9ae1e50b2dfb413eaf858120b4dfdfc8 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2932eb4e2874ba6d4702d5e07179ded99aa51400","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-06 10:18:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c38376aefc072651e81342679e06b57683dd5677","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 10:19:57.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"1bacd37d92982ed369c1ea4edf9dc9647b3821e7","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-06 11:27:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aa1a2da59f67a2aa5a36018017ec3b186375141d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-06 16:08:49.000000000","message":"Patch Set 1: Code-Review-1\n\nIs it good to release without reaching a conclusion on https://review.opendev.org/#/c/711410/ ? (sorry I suggested the initial change to go in 2.2.0)","accounts_in_message":[],"_revision_number":1},{"id":"c28bb11e8eaebf2fd5e5f63cf7f7e22306a4e743","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-07 18:37:47.000000000","message":"Patch Set 1:\n\n\u003e Is it good to release without reaching a conclusion on\n \u003e https://review.opendev.org/#/c/711410/ ? (sorry I suggested the\n \u003e initial change to go in 2.2.0)\n\nSure, we can wait for this or release now and maybe later release 2.2.1 with this additional patch when it will be finally decided what to do with those logs.","accounts_in_message":[],"_revision_number":1},{"id":"94d20fa860b972c62af38e8dff0db441e220ae30","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-08 17:55:58.000000000","message":"Patch Set 1: Code-Review-1\n\nnow I would like to wait also for https://review.opendev.org/711856","accounts_in_message":[],"_revision_number":1},{"id":"0787390b535262438d47a7818d72b5bf3627965b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-08 17:56:05.000000000","message":"Patch Set 1: -Code-Review Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"56cbd376fe829ad23bc862cae2947a38b3a339a2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-10 08:28:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"470a1bc901c5abba1e979f2ffc3a4d154b5255e6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-10 08:29:01.000000000","message":"Patch Set 2:\n\nBoth patches mentioned in the comments above are now merged so we should be ready to go with this new release now.","accounts_in_message":[],"_revision_number":2},{"id":"4f65298e88ab9e7dd3d69bcdb0bdb22c3e24f854","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 08:29:42.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"7827d37cb3b1f5af3d58ed29bed8992a32f823f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 08:39:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797965d1ae9d42c5887dcc08db98f753 : SUCCESS in 10m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/275d5df5d8a44efea91d13edd8d24077 : SUCCESS in 6m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/de74cec6460a4180948a258a02e3e048 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"00f1fe05e2011728cd4ff21d2c17f07acbc28114","author":{"_account_id":14232,"name":"Igor Malinovskiy","email":"u.glide@gmail.com","username":"imalinovskiy"},"date":"2020-03-10 08:40:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1f282c2f94f6b8d0688785694629319a734d5d4e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-10 09:07:08.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"c46714fbb334dd294915f0d61f5e6525569b093b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-10 10:36:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b0a5e352401cea363b8da423a88322c8586fb18e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-10 13:48:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a520e5d034dce92e13c96e6d34072bcc22c5a6e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 13:49:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf42643b11798f119873fa438e2ce9f63344c006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 14:10:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c077a2bcf342cf8473ff81cc075f7f : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd9add493014ce7ad3fd55b4a0e5978 : SUCCESS in 13m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38e69c00be944e868c75438826dd5a71 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6ab204ba8c3f3bbeb82994b8da0e262321c08d21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 14:10:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb4bbbd3ba931a9521694f1323bc3979ee6e5c71","revisions":{"f930beb7015c29d1d010ebeaaaac270b8a4937b5":{"kind":"REWORK","_number":1,"created":"2020-03-06 09:57:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/00/711600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/711600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/711600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/711600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/711600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/711600/1"}}},"commit":{"parents":[{"commit":"5d392090b3d7d6bb2e85bf9ffca2481c4487f13a","subject":"Merge \"[vitrage] Transition Rocky to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d392090b3d7d6bb2e85bf9ffca2481c4487f13a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-06 09:55:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-06 09:55:59.000000000","tz":60},"subject":"Release neutron-lib 2.2.0","message":"Release neutron-lib 2.2.0\n\nNew changes:\n - rbac-address-scope api extension\n - rbac-subnetpool api extension\n - is_sort_key added to vpnaas attribute maps\n - move warnings of using old and new enginefacade to be debug logs\n\nChange-Id: I95d99f9aa1eb850ab8d6d2c1223d92fa8b58dea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f930beb7015c29d1d010ebeaaaac270b8a4937b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f930beb7015c29d1d010ebeaaaac270b8a4937b5"}]},"branch":"refs/heads/master"},"fb4bbbd3ba931a9521694f1323bc3979ee6e5c71":{"kind":"REWORK","_number":2,"created":"2020-03-10 08:28:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/00/711600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/711600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/711600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/711600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/711600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/711600/2"}}},"commit":{"parents":[{"commit":"5d392090b3d7d6bb2e85bf9ffca2481c4487f13a","subject":"Merge \"[vitrage] Transition Rocky to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d392090b3d7d6bb2e85bf9ffca2481c4487f13a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-06 09:55:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-10 08:28:34.000000000","tz":0},"subject":"Release neutron-lib 2.2.0","message":"Release neutron-lib 2.2.0\n\nNew changes:\n - rbac-address-scope api extension\n - rbac-subnetpool api extension\n - is_sort_key added to vpnaas attribute maps\n - move warnings of using old and new enginefacade to be debug logs\n\nChange-Id: I95d99f9aa1eb850ab8d6d2c1223d92fa8b58dea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb4bbbd3ba931a9521694f1323bc3979ee6e5c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb4bbbd3ba931a9521694f1323bc3979ee6e5c71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
