)]}'
{"id":"openstack%2Freleases~712126","triplet_id":"openstack%2Freleases~master~Ib212da7c7f76d858331a54dd12c2a8e9d2cec337","project":"openstack/releases","branch":"master","topic":"magnum","hashtags":[],"change_id":"Ib212da7c7f76d858331a54dd12c2a8e9d2cec337","subject":"Release magnum train 9.3.0","status":"MERGED","created":"2020-03-10 16:52:00.000000000","updated":"2020-03-31 14:21:04.000000000","submitted":"2020-03-31 14:21:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712126-1585664464963-93ed6149","meta_rev_id":"f253bbd03f5d5775d9c2952dcf0a98fb972d9c7f","_number":712126,"virtual_id_number":712126,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-03-31 10:35:01.000000000","_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 14:21:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":1,"date":"2020-03-30 21:54:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":2,"date":"2020-03-31 14:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-03-31 11:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":1,"date":"2020-03-31 14:07:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"date":"2020-03-31 14:05:40.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-30 21:55:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-10 16:56:38.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2020-03-10 20:20:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-30 21:54:11.000000000","updated_by":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"reviewer":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"state":"REVIEWER"},{"updated":"2020-03-31 11:20:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-31 14:05:40.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-03-31 14:07:38.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-03-31 14:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa93d2e925350a2c9d7e2452962435afd4fb9bb1","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-10 16:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"413998b54e85da9aac11622d0c24f9e3e5364419","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-10 16:53:02.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4d34fc0b30c193d528f176e2e92403c82d474ddf","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-10 16:55:49.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"eeba8746053fb03d9c06dbe4d0e1a45fafbafd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 17:12:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b6c31d27a84a5d8cbfc323c9756701 : SUCCESS in 10m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6855dae0aa4d4812bd0c4129c19d5d42 : SUCCESS in 6m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dd5da68a78fd4e9c9f22bb8caa7eca0c : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":3},{"id":"53d85d0e0d635e7bc1801dc8999e8964d26ae4e5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-10 20:20:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1cc8856b18046b42f4fc251012a5881d74e83a11","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-11 09:42:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a6f1133a4b947313eba40da3ff138617055f8b43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 09:43:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9f0a05e37be0ae575161c43820fb08aef5bf5505","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-11 09:43:36.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"4f805eaf5baad681c5181b0977349fc875e3eee0","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-11 09:45:26.000000000","message":"Patch Set 3: Workflow-1\n\nIn the weekly meeting, there are a couple more fixes that need to be merged before we cut this release.","accounts_in_message":[],"_revision_number":3},{"id":"996d2b3a5ba90370c3ddb51c77b12bba3becef17","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-11 09:46:24.000000000","message":"Patch Set 3:\n\nWaiting for validation from PTL (Feilong Wang) or liaison (Spyros Trigazis) first","accounts_in_message":[],"_revision_number":3},{"id":"7e63a0792572fdc6572624bfb22000c897e78d2f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-11 09:47:57.000000000","message":"Patch Set 3: Code-Review+1\n\n\u003e In the weekly meeting, there are a couple more fixes that need to\n \u003e be merged before we cut this release.\n\nOk so I move to +1 only for now, thanks for the heads up.","accounts_in_message":[],"_revision_number":3},{"id":"f3035a8d568126b3c868b13d688f03464d7f1006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 09:55:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2201b9fe89654fc4a9685a38b86f80c3 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538ff8dfccd747d8947f1879fa0fa077 : SUCCESS in 11m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/113e775c3728425db45991814018efce : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":3},{"id":"625db9e51f660ee914c233b04a4550132e8ce845","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-30 16:44:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5011673afccfa6245f944d8bfca22c7b47c1777e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 17:19:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9791c5bde52c4d838ba07ce532914204 : SUCCESS in 12m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4122c74b103b4c228319c6e989f9d29c : SUCCESS in 9m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e3af74c7dd2540ac9ac2c25a83a26e23 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":4},{"id":"cb9fe5aa085486eaa7293f9c68db37315a4c25f6","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-03-30 21:54:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f93401af8db0995b376c24e8496f423ec02341ad","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 21:55:47.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"696a1690d43b6435d65ca3899019ba7fdb6abab2","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-31 10:35:01.000000000","message":"Patch Set 4:\n\n@Spyros are you also happy with this?","accounts_in_message":[],"_revision_number":4},{"id":"c376309318dc93b8d375955348037436119c5d3b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-31 11:20:59.000000000","message":"Patch Set 4: Code-Review+2\n\nNew features have been added in the changes embed in this release, this is something that we want to avoid [1], however magnum doesn\u0027t subscribe to the stable branches policy [2] so I think it\u0027s ok to release new features and a minor version.\n\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#active-maintenance\n[2] https://github.com/openstack/governance/blob/master/reference/projects.yaml#L1080,L1111","accounts_in_message":[],"_revision_number":4},{"id":"6399556c08251c260272e27a42fc1b911dc7c2f5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-31 14:07:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"63fe1b9da2755766934d2175d02d7bc26b4f5e83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:07:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"56ce6a56500cf1b68718a0072ed3ce02e1e4b2b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:21:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"892bef79bda497f946436f878af41aea69cb458d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:21:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b693e209b48744eebd763b2eebf19c71 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7cc9690d2e49c299dd4e1099cd8da5 : SUCCESS in 12m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0941f56006bb46aa9f695b14651eff3e : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f2f56c3281dde6ed48d44be92001cdb6b24c63a0","revisions":{"c656d5dfbdf66c9b2b905a04d3c3110c97e856b6":{"kind":"REWORK","_number":1,"created":"2020-03-10 16:52:00.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/26/712126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/712126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/712126/1"}}},"commit":{"parents":[{"commit":"ba138eb10bea0a9df5b5060f98e2db643dba1740","subject":"Release magnum 9.2.0 for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba138eb10bea0a9df5b5060f98e2db643dba1740"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-10 16:51:35.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-10 16:51:35.000000000","tz":0},"subject":"Release magnum 9.3.0 for Train","message":"Release magnum 9.3.0 for Train\n\nChange-Id: Ib212da7c7f76d858331a54dd12c2a8e9d2cec337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c656d5dfbdf66c9b2b905a04d3c3110c97e856b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c656d5dfbdf66c9b2b905a04d3c3110c97e856b6"}]},"branch":"refs/heads/master"},"e6ab509eee74080b53761aef943839a2fd7d4ebc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-10 16:53:02.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/26/712126/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/712126/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/712126/2"}}},"commit":{"parents":[{"commit":"ba138eb10bea0a9df5b5060f98e2db643dba1740","subject":"Release magnum 9.2.0 for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba138eb10bea0a9df5b5060f98e2db643dba1740"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-10 16:51:35.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-10 16:52:52.000000000","tz":0},"subject":"Release magnum 9.3.0 for Train","message":"Release magnum 9.3.0 for Train\n\nChange-Id: Ib212da7c7f76d858331a54dd12c2a8e9d2cec337\nSigned-off-by: Bharat Kunwar \u003cbrtknr@bath.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6ab509eee74080b53761aef943839a2fd7d4ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6ab509eee74080b53761aef943839a2fd7d4ebc"}]},"branch":"refs/heads/master"},"512ee73f669f2f7b300511087e66a55829524335":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-03-10 16:55:49.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/26/712126/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/712126/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/712126/3"}}},"commit":{"parents":[{"commit":"ba138eb10bea0a9df5b5060f98e2db643dba1740","subject":"Release magnum 9.2.0 for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba138eb10bea0a9df5b5060f98e2db643dba1740"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-10 16:51:35.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-10 16:55:21.000000000","tz":0},"subject":"Release magnum train 9.3.0","message":"Release magnum train 9.3.0\n\nChange-Id: Ib212da7c7f76d858331a54dd12c2a8e9d2cec337\nSigned-off-by: Bharat Kunwar \u003cbrtknr@bath.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/512ee73f669f2f7b300511087e66a55829524335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/512ee73f669f2f7b300511087e66a55829524335"}]},"branch":"refs/heads/master"},"f2f56c3281dde6ed48d44be92001cdb6b24c63a0":{"kind":"REWORK","_number":4,"created":"2020-03-30 16:44:52.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/26/712126/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/712126/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/712126/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/712126/4"}}},"commit":{"parents":[{"commit":"ba138eb10bea0a9df5b5060f98e2db643dba1740","subject":"Release magnum 9.2.0 for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba138eb10bea0a9df5b5060f98e2db643dba1740"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-10 16:51:35.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-30 16:44:50.000000000","tz":0},"subject":"Release magnum train 9.3.0","message":"Release magnum train 9.3.0\n\nChange-Id: Ib212da7c7f76d858331a54dd12c2a8e9d2cec337\nSigned-off-by: Bharat Kunwar \u003cbrtknr@bath.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2f56c3281dde6ed48d44be92001cdb6b24c63a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2f56c3281dde6ed48d44be92001cdb6b24c63a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
