)]}'
{"id":"openstack%2Freleases~713941","triplet_id":"openstack%2Freleases~master~Ib63f2b63a8e8f269fe18efe8e0d20771eec75c47","project":"openstack/releases","branch":"master","topic":"rocky-em","hashtags":[],"change_id":"Ib63f2b63a8e8f269fe18efe8e0d20771eec75c47","subject":"Create stable/rocky branch for tripleo-ipsec","status":"MERGED","created":"2020-03-19 17:24:56.000000000","updated":"2020-03-19 21:39:24.000000000","submitted":"2020-03-19 21:39:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713941-1584653964796-b7a9cff9","meta_rev_id":"551d866fa46e5692ac48e9bc1632e26f250c0e06","_number":713941,"virtual_id_number":713941,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 21:39:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-19 21:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-19 21:23:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 21:39:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b68a55db4d29948bcc54a4ee2d441730ef7d6ad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-19 17:24:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d239f24f9d4b2010c7bed442eb4dc5b905be5ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 17:49:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed19c657309847c3ac4efa5fb01968ae : SUCCESS in 11m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cdae270bbacb433c9f8664fac5861a00 : SUCCESS in 9m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3a9b272812334aac8150150950104971 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c1fbbaab09c3d21fd5c4329a2cfc567172b53442","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-19 21:23:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nJust needed to close out this branch into rocky-em.","accounts_in_message":[],"_revision_number":1},{"id":"0bc71f3cfac62d89711113c37dbae4e297f7696a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:24:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fedc10c3708302a90e13227b482a518d3d3a55cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:39:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"68ac8a10728227d5d0b84181791f2943efca6b3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:39:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ec26fcf3fdf4a34acac694852ac9bca : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f181262dd03481f885e1f8101d604fe : SUCCESS in 12m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/610b8330e08e4023831a67ab09e370bf : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aeef556e069912d0dd939ac943fde697473d8342","revisions":{"aeef556e069912d0dd939ac943fde697473d8342":{"kind":"REWORK","_number":1,"created":"2020-03-19 17:24:56.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/713941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/713941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/713941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/713941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/713941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/713941/1"}}},"commit":{"parents":[{"commit":"36a88565bbdfec86932f2c845a8fde64b802f929","subject":"Merge \"[Telemetry] Transition Rocky to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/36a88565bbdfec86932f2c845a8fde64b802f929"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-19 17:19:46.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-19 17:20:35.000000000","tz":-300},"subject":"Create stable/rocky branch for tripleo-ipsec","message":"Create stable/rocky branch for tripleo-ipsec\n\nChange-Id: Ib63f2b63a8e8f269fe18efe8e0d20771eec75c47\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aeef556e069912d0dd939ac943fde697473d8342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aeef556e069912d0dd939ac943fde697473d8342"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
