)]}'
{"id":"openstack%2Freleases~714155","triplet_id":"openstack%2Freleases~master~If4add04307cf459c92b8ee563a3e3e4a512e2fdc","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"If4add04307cf459c92b8ee563a3e3e4a512e2fdc","subject":"Release cloudkitty deliverables for stable/rocky","status":"MERGED","created":"2020-03-20 15:57:06.000000000","updated":"2020-03-20 16:43:48.000000000","submitted":"2020-03-20 16:43:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714155-1584722628258-332a2ac9","meta_rev_id":"6f00d126eb64b01c7f5efb083715f1e188a76e21","_number":714155,"virtual_id_number":714155,"owner":{"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 16:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-20 16:05:24.000000000","_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-20 16:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-20 16:27:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 16:27:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-20 16:43:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01f6ddf7de8ff78a1dfc97faf05086671217477c","author":{"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"},"date":"2020-03-20 15:57:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6efdf9ee18a492a23cf8620f241a30301dd3df2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-20 16:02:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nIssue with the dashboard version number.\n\nYou can use \"tools/new_release.sh rocky cloudkitty-dashboard bugfix\" to make these easier for future releases.","accounts_in_message":[],"_revision_number":1},{"id":"261e849c8537ff2e70ffcefbb34bbb6054ff0523","author":{"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"},"date":"2020-03-20 16:04:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bd9601ceb3f19ca685cda0a7f146bd82ed69861e","author":{"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"},"date":"2020-03-20 16:05:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43bda4d217c3008494df58c8f990f7a7c61f93a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:22:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02772488782f43d3923311a8341f78d5 : SUCCESS in 10m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fac2a3a2381c4bb6851200d8fec4313a : SUCCESS in 7m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5fab90e69bbb4bf1a974b87b3dbbc518 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8bcf81c5650254a869d5e4f1a049b79f5392ca57","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-20 16:27:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nJust minor changes being released, and past our deadline for EM, so going to push this through so we can close things out.","accounts_in_message":[],"_revision_number":2},{"id":"e474aa63412b529a3f00ee360d20e82ad536bebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:27:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a3f1c46ea07e7ce39581c1a86ff3ee6b5d6ba65a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:43:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae46ad91412d483fbe87559091666a6e : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520349ce2788445a9181736394f3ac84 : SUCCESS in 11m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66ef2a62e3a5415b89743bd78907ed50 : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6f00d126eb64b01c7f5efb083715f1e188a76e21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:43:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1002abbc240c866756fc8c1ef09a7528f405fbe4","revisions":{"c51376f330bcdeb2c8c98be66b9c11221de5061f":{"kind":"REWORK","_number":1,"created":"2020-03-20 15:57:06.000000000","uploader":{"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"},"ref":"refs/changes/55/714155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/714155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/714155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/714155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/714155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/714155/1"}}},"commit":{"parents":[{"commit":"f7dfd832c5ff5d12b090eb553db0927f1b0c5e5f","subject":"Merge \"[tripleo] Transition Rocky to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7dfd832c5ff5d12b090eb553db0927f1b0c5e5f"}]}],"author":{"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","date":"2020-03-20 15:53:44.000000000","tz":60},"committer":{"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","date":"2020-03-20 15:57:04.000000000","tz":60},"subject":"Release cloudkitty deliverables for stable/rocky","message":"Release cloudkitty deliverables for stable/rocky\n\nChange-Id: If4add04307cf459c92b8ee563a3e3e4a512e2fdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c51376f330bcdeb2c8c98be66b9c11221de5061f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c51376f330bcdeb2c8c98be66b9c11221de5061f"}]},"branch":"refs/heads/master"},"1002abbc240c866756fc8c1ef09a7528f405fbe4":{"kind":"REWORK","_number":2,"created":"2020-03-20 16:04:37.000000000","uploader":{"_account_id":29503,"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","username":"jferrieu"},"ref":"refs/changes/55/714155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/714155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/714155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/714155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/714155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/714155/2"}}},"commit":{"parents":[{"commit":"f7dfd832c5ff5d12b090eb553db0927f1b0c5e5f","subject":"Merge \"[tripleo] Transition Rocky to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7dfd832c5ff5d12b090eb553db0927f1b0c5e5f"}]}],"author":{"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","date":"2020-03-20 15:53:44.000000000","tz":60},"committer":{"name":"Justin Ferrieu","email":"jferrieu@objectif-libre.com","date":"2020-03-20 16:04:32.000000000","tz":0},"subject":"Release cloudkitty deliverables for stable/rocky","message":"Release cloudkitty deliverables for stable/rocky\n\nChange-Id: If4add04307cf459c92b8ee563a3e3e4a512e2fdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1002abbc240c866756fc8c1ef09a7528f405fbe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1002abbc240c866756fc8c1ef09a7528f405fbe4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
