)]}'
{"id":"openstack%2Freleases~714405","triplet_id":"openstack%2Freleases~master~I73ded29fdbf1c5c6e0fa82d48b5fd5fadc401689","project":"openstack/releases","branch":"master","topic":"release-neutron-tempest-plugin","hashtags":[],"change_id":"I73ded29fdbf1c5c6e0fa82d48b5fd5fadc401689","subject":"Release neutron-tempest-plugin 1.0.0","status":"MERGED","created":"2020-03-23 09:44:21.000000000","updated":"2020-03-23 14:48:39.000000000","submitted":"2020-03-23 14:48:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714405-1584974919509-81340386","meta_rev_id":"0e5630674b449ca258fd7097a4f266c7c72845af","_number":714405,"virtual_id_number":714405,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 14:48:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-23 14:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-23 14:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-23 14:37:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 14:32:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-23 14:48:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"153674987edceaa6241531bea620fe78cb56df02","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-23 09:44:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea3e14a26b25461f6a8aac0f9de775835cdb934d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 09:52:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3eda28a94643039664173e4476ff55 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e5433f3b6ba43789db5cf98520e12e1 : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b5eb2870531e4811824fbaf72429c5da : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"90aef61619cc5e0bfd63157f6ec874c0d0046039","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-23 14:32:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b5271efa49d6a9cc589e5b8045aad44d7e8ab54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:33:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9483576955d0d22508ba545f76aef33bb739dda2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:37:30.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"c0efbbb1b15ac6e05455f36257165e0a4553c3b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:48:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"371d5616d72738dd40e4a975adc25112c7752e3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:48:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdc337cb4924add93faa8e02c5691b5 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b20a424767e4ecd8065cb1611215ddd : SUCCESS in 13m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/118ed80528da4c6b8099f74965c2df56 : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"105d95a3db5df9e1b03506cf8f6c6cea0d83dc1d","revisions":{"105d95a3db5df9e1b03506cf8f6c6cea0d83dc1d":{"kind":"REWORK","_number":1,"created":"2020-03-23 09:44:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/05/714405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/714405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/714405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/714405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/714405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/714405/1"}}},"commit":{"parents":[{"commit":"cd5f2011a7646d64e2c03fafce56a0fef934314a","subject":"oslo.policy 3.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd5f2011a7646d64e2c03fafce56a0fef934314a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-23 09:42:25.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-23 09:44:10.000000000","tz":60},"subject":"Release neutron-tempest-plugin 1.0.0","message":"Release neutron-tempest-plugin 1.0.0\n\nThis new version has dropped support for Python 2 and because\nof that there is major version bump.\nIt also has some new test cases and bug fixes which may be useful e.g.\nin downstream testing.\n\nChange-Id: I73ded29fdbf1c5c6e0fa82d48b5fd5fadc401689\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/105d95a3db5df9e1b03506cf8f6c6cea0d83dc1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/105d95a3db5df9e1b03506cf8f6c6cea0d83dc1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
