)]}'
{"id":"openstack%2Freleases~714586","triplet_id":"openstack%2Freleases~master~I3e9c5e64d4ca2b49bc4ce2689834a62d6ed71b42","project":"openstack/releases","branch":"master","topic":"release","hashtags":[],"change_id":"I3e9c5e64d4ca2b49bc4ce2689834a62d6ed71b42","subject":"Bump mistral-lib version to 2.0.0","status":"MERGED","created":"2020-03-24 09:22:43.000000000","updated":"2020-03-24 10:34:37.000000000","submitted":"2020-03-24 10:34:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714586-1585046077253-046e596a","meta_rev_id":"606f45da8cacb21296102c96c347049055d20668","_number":714586,"virtual_id_number":714586,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 10:34:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-24 10:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-24 10:31:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-24 10:22:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-24 10:33:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 10:22:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-24 10:31:50.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-03-24 10:34:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c97a5a985a61f9d7637e0c4e23e293dd52eee42e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-24 09:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"575a9055232a1138fff421937f74620f1a803222","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-24 09:34:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e48a53071bffe77fb0cfebc3f50bb122b31d24df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 09:48:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1a4db2198949b8bb8b69026a616a88 : SUCCESS in 9m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9838c7911f1f4c99bb8576dd23e8fe24 : SUCCESS in 6m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3bab605457d8488eb3ec9b1d159c8c38 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"6f16260b0aaacb9ffdc8abb37dc01858d8455d3c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-24 09:54:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"572026401e23338e127383016b0d3e16e83c1bfc","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-24 09:55:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bd9db7cb0077b756511d77defcc6123e30ad3064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 10:05:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb7545d21b043ee8691c86744521d3d : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3fa5631cabb47c1b13ed5709003d282 : SUCCESS in 5m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/24d00f0b205244cd9b893e4ffd6deca1 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9c78f4c69c046c413cd02436f2f7d1da1bf5069f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-24 10:22:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1b99903d3b8a381fe84f9efd92bb1c8425a120bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 10:22:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c859214eb538fd9cf6878120e67083a560105560","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-24 10:31:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b0aab454d0a46239d06c45b21d3d24af74dedef0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 10:33:03.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"2fbae18cb77ed181ac4da85d7d3a78167b4cc5fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-24 10:34:04.000000000","message":"Patch Set 3:\n\nSorry, I had just realized I didn\u0027t wait for PTL ack. Thanks Renat.","accounts_in_message":[],"_revision_number":3},{"id":"7e0b14c0e2702326182c5d7a41957dc46ee4a72c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 10:34:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9070b9faff4557b974fc1a053fb945 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4463716343342e2b7cb872961e02f60 : SUCCESS in 10m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e88b4dfdb18f400caf91e46bfa3f848c : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"99106e7ebfaece5f1fe463a1d82f91e51062d592","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 10:34:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe1cece487f2b26cb3ed68ea501e102bb3c47c47","revisions":{"223eb47afbcbe3ec6ff710d38284dcbcd0db4642":{"kind":"REWORK","_number":1,"created":"2020-03-24 09:22:43.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/86/714586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/714586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/714586/1"}}},"commit":{"parents":[{"commit":"b5b9841e619554ca77614c426fbcf106ada3bb68","subject":"Merge \"pymod2pkg: release 0.23\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5b9841e619554ca77614c426fbcf106ada3bb68"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-24 09:22:00.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-24 09:22:00.000000000","tz":120},"subject":"Bump mistral-lib version to 1.4.1","message":"Bump mistral-lib version to 1.4.1\n\nChange-Id: I3e9c5e64d4ca2b49bc4ce2689834a62d6ed71b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/223eb47afbcbe3ec6ff710d38284dcbcd0db4642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/223eb47afbcbe3ec6ff710d38284dcbcd0db4642"}]},"branch":"refs/heads/master"},"149254ae1b1cb9d9b008e70852eccdee502a16b4":{"kind":"REWORK","_number":2,"created":"2020-03-24 09:34:32.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/86/714586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/714586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/714586/2"}}},"commit":{"parents":[{"commit":"b5b9841e619554ca77614c426fbcf106ada3bb68","subject":"Merge \"pymod2pkg: release 0.23\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5b9841e619554ca77614c426fbcf106ada3bb68"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-24 09:22:00.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-24 09:34:19.000000000","tz":120},"subject":"Bump mistral-lib version to 1.4.1","message":"Bump mistral-lib version to 1.4.1\n\nChange-Id: I3e9c5e64d4ca2b49bc4ce2689834a62d6ed71b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/149254ae1b1cb9d9b008e70852eccdee502a16b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/149254ae1b1cb9d9b008e70852eccdee502a16b4"}]},"branch":"refs/heads/master"},"fe1cece487f2b26cb3ed68ea501e102bb3c47c47":{"kind":"REWORK","_number":3,"created":"2020-03-24 09:55:49.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/86/714586/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/86/714586/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/86/714586/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/86/714586/3"}}},"commit":{"parents":[{"commit":"b5b9841e619554ca77614c426fbcf106ada3bb68","subject":"Merge \"pymod2pkg: release 0.23\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5b9841e619554ca77614c426fbcf106ada3bb68"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-24 09:22:00.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-24 09:55:19.000000000","tz":120},"subject":"Bump mistral-lib version to 2.0.0","message":"Bump mistral-lib version to 2.0.0\n\nChange-Id: I3e9c5e64d4ca2b49bc4ce2689834a62d6ed71b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe1cece487f2b26cb3ed68ea501e102bb3c47c47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe1cece487f2b26cb3ed68ea501e102bb3c47c47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
