)]}'
{"id":"openstack%2Freleases~714805","triplet_id":"openstack%2Freleases~master~I12c862d7cd840a091e980f32de3178fbabfa5dac","project":"openstack/releases","branch":"master","topic":"cinderproject-mar-2020","hashtags":[],"change_id":"I12c862d7cd840a091e980f32de3178fbabfa5dac","subject":"Release python-cinderclient 4.2.2","status":"MERGED","created":"2020-03-25 01:56:09.000000000","updated":"2020-03-25 14:27:08.000000000","submitted":"2020-03-25 14:27:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714805-1585146428605-4f1731f4","meta_rev_id":"101ed1745d3591ac46024390564ef0d0ed57e656","_number":714805,"virtual_id_number":714805,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 14:27:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-25 14:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-25 10:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-25 14:13:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-25 01:57:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 10:20:32.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-03-25 14:13:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-25 14:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11ba2c7482bb49ef742014ed43d6db006105ae88","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-25 01:56:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cad2f33f9587ac28c760b3709996bdefc07de4ae","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 01:57:18.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"f251672a98200a5e857ea60e0cdf4528b7de56f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 02:07:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2892e28d75141408ba2e8bfb51eedfe : SUCCESS in 9m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35c4a9d1597d4d1aae9733cd69667995 : SUCCESS in 6m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/104602c14f1d4b0b8256588b0760b3a9 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"70c1a8a106e18db60c68a80202d3ed7272edf15c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-25 10:20:32.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm.","accounts_in_message":[],"_revision_number":1},{"id":"4e86add66afa573f4265bd63e30427cda065ef36","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-25 11:42:24.000000000","message":"Topic changed from release-cinderclient-4.2.2 to cinderproject-mar-2020","accounts_in_message":[],"_revision_number":1},{"id":"7e87662bd5b09c46e5ebb20b929d6518b2a8a9e1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-25 14:13:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nA couple good bugfixes.","accounts_in_message":[],"_revision_number":1},{"id":"4ecc8d7ae4229eb98fdd030da09554b4b197db5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 14:14:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5057b0ec8744c54421ac0a621a8b1410de2bd0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 14:27:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70a2af91a9b47889ceb16a0dac0f1d4 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38694e8e854431b93350820cebaf0f0 : SUCCESS in 10m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/337fd5c6a617445e8590315911e32989 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"69ca6eb4312fca1909223d7a94773d7d3747c9d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 14:27:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"872a73a9dc351cad2b6c56fcc0b6264360aec938","revisions":{"872a73a9dc351cad2b6c56fcc0b6264360aec938":{"kind":"REWORK","_number":1,"created":"2020-03-25 01:56:09.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/05/714805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/714805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/714805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/714805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/714805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/714805/1"}}},"commit":{"parents":[{"commit":"15267522fb526301c5e5f5cd363db4e6b76720e8","subject":"Release python-cinderclient 5.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15267522fb526301c5e5f5cd363db4e6b76720e8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-25 01:55:16.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-25 01:55:16.000000000","tz":-240},"subject":"Release python-cinderclient 4.2.2","message":"Release python-cinderclient 4.2.2\n\nRelease from stable/stein\n\nChange-Id: I12c862d7cd840a091e980f32de3178fbabfa5dac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/872a73a9dc351cad2b6c56fcc0b6264360aec938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/872a73a9dc351cad2b6c56fcc0b6264360aec938"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
