)]}'
{"id":"openstack%2Freleases~714826","triplet_id":"openstack%2Freleases~master~If9eedc7fd3881c46bfd77d2f87339157c371995d","project":"openstack/releases","branch":"master","topic":"horizon","hashtags":[],"change_id":"If9eedc7fd3881c46bfd77d2f87339157c371995d","subject":"Release horizon 18.2.0 (ussuri)","status":"MERGED","created":"2020-03-25 04:16:07.000000000","updated":"2020-03-25 10:35:31.000000000","submitted":"2020-03-25 10:35:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714826-1585132531275-c63a599f","meta_rev_id":"54923adf807ade03f6cee8d3f55b42de2d7e2260","_number":714826,"virtual_id_number":714826,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-25 04:53:35.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 10:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-25 10:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-25 10:23:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-25 04:17:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 10:23:21.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-03-25 10:35:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e428cc20ded067053f92ace1f0f99dfb7d89143","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-25 04:16:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"002d8a68d847fd4fdbead726a1d4458d5549c647","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 04:17:08.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"634b0f8656d9230019b021bd8c1e7ad0437053d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 04:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2498fd7eee4499807c0ad618a9077c : SUCCESS in 8m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57389b314cf44dafb1b38643f043055c : SUCCESS in 9m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/733543a9385a4b18975339fde5dd7f28 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9ee6237991dc1815af5b77637edcde4c4fb02f5c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-25 04:53:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39c6874d6f18110c03ddd31bd17c1946bba0b409","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-25 10:22:17.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"ca59698c3d2d529a64c6a8a2c73e4e845eb7bd4c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-25 10:23:21.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c326b85bcb4bbab2101c34b5ffda759a5f97df2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 10:23:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ee4daf623f078e1e1bf006483e2518dc74ca12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 10:35:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0310fc6408dd4493862098a2479f42f2 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a490bdfb9b1246019c9facbab7d77ce9 : SUCCESS in 10m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c451ff93c7314d05b0e06ac38ec6a43c : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a58bda6a750c3a14dacadf90d3b2bdc20d95aa36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 10:35:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f1b99a8984f45ec5b8605fd02588816a90b863c","revisions":{"4f1b99a8984f45ec5b8605fd02588816a90b863c":{"kind":"REWORK","_number":1,"created":"2020-03-25 04:16:07.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/26/714826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/714826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/714826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/714826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/714826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/714826/1"}}},"commit":{"parents":[{"commit":"3c47c2557b807fb11f472f585dcbf587e5648e14","subject":"Merge \"Add Oslo feature freeze to Victoria schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c47c2557b807fb11f472f585dcbf587e5648e14"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-03-25 03:54:02.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-03-25 03:59:28.000000000","tz":540},"subject":"Release horizon 18.2.0 (ussuri)","message":"Release horizon 18.2.0 (ussuri)\n\npyscss is now maintained by support of horizon team so we can move back to\npyscss from pyscss2 (fork). It looks better to cut a new release of horizon\nto avoid pyscss2 usage including lower-constraints in horizon plugins.\n\nChange-Id: If9eedc7fd3881c46bfd77d2f87339157c371995d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f1b99a8984f45ec5b8605fd02588816a90b863c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f1b99a8984f45ec5b8605fd02588816a90b863c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
