)]}'
{"id":"openstack%2Freleases~715274","triplet_id":"openstack%2Freleases~master~I8ebbb2ca65c891304afdc17d6acc60a4c58608db","project":"openstack/releases","branch":"master","topic":"xstatic-releases","hashtags":[],"change_id":"I8ebbb2ca65c891304afdc17d6acc60a4c58608db","subject":"Release xstatic-bootstrap-scss 3.4.1.0","status":"ABANDONED","created":"2020-03-26 17:20:55.000000000","updated":"2020-03-27 10:20:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"445ebd6d844acbcfc2b183d49c309ade43227c32","_number":715274,"virtual_id_number":715274,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-26 17:37:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2020-03-26 18:58:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-03-26 17:20:55.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-26 17:20:55.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-03-26 17:37:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e66ea5ea39628a125fa62628dc354c13162d089","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-26 17:20:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bb69925c86f4454a5849a6ce994373eeb3d8c47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 17:37:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e45d0736909d481981c364326af79260 : SUCCESS in 13m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/099cb8cb49914abdb2eedc79e8a63074 : SUCCESS in 7m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/85de4b3abbf44207b61eb4ed2f708d7c : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":1},{"id":"92476e91fd340f63f29208bbe0b55343307865c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-26 18:58:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b48d6949565b546e788af4c178ae02622c5a414a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-27 10:20:14.000000000","message":"Abandoned\n\nWas created for testing purposes","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"486902a3edcdd1dfcbc1ec71eefa75570c396a27","revisions":{"486902a3edcdd1dfcbc1ec71eefa75570c396a27":{"kind":"REWORK","_number":1,"created":"2020-03-26 17:20:55.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/74/715274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/715274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/715274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/715274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/715274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/715274/1"}}},"commit":{"parents":[{"commit":"97860919df69812e2a90792924584be5e3a9a031","subject":"Merge \"Use governance_consistency.py for MembershipFreeze\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97860919df69812e2a90792924584be5e3a9a031"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-03-24 15:46:04.000000000","tz":540},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-26 17:20:21.000000000","tz":-300},"subject":"Release xstatic-bootstrap-scss 3.4.1.0","message":"Release xstatic-bootstrap-scss 3.4.1.0\n\nChange-Id: I8ebbb2ca65c891304afdc17d6acc60a4c58608db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/486902a3edcdd1dfcbc1ec71eefa75570c396a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/486902a3edcdd1dfcbc1ec71eefa75570c396a27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
