)]}'
{"id":"openstack%2Freleases~715282","triplet_id":"openstack%2Freleases~master~I899b5cb74e2902c8e5d37a244ebefaa98dbf30e1","project":"openstack/releases","branch":"master","topic":"rocky-em","hashtags":[],"change_id":"I899b5cb74e2902c8e5d37a244ebefaa98dbf30e1","subject":"Release new rocky version for heat repos","status":"MERGED","created":"2020-03-26 17:44:58.000000000","updated":"2020-04-02 11:06:22.000000000","submitted":"2020-04-02 11:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715282-1585825582649-e8a3316c","meta_rev_id":"a2efb5f78c3b9382f81320405f47cead41b94a23","_number":715282,"virtual_id_number":715282,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-02 07:12:55.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-03-27 17:43:55.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 11:06:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-02 10:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-04-02 09:44:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 10:35:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-27 10:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 17:43:55.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-02 07:12:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-02 09:44:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-02 10:35:05.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-02 11:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9ffe45331813acb990683ddbd18bc64cbeedf57","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-03-26 17:44:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d1a477cd340bccb0b2681f47f3aeb9e91fed08d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 17:47:02.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"c6dff2746462d9f5a338f637a91c75029ca9e618","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-26 17:50:02.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"ccc2ab09d620e53175157dc5910d75bcf5166c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 18:04:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198b1e9becd04bf9acb25aa4a727e367 : SUCCESS in 10m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aed1100e219b488bbd8ba4ba25e4756f : FAILURE in 17m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/280faf3cd5e740e49cddb15fecf78ebe : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"480b721d0b504ed729563264fc75308435f79786","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-03-26 18:12:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ecebb78b54df21ae3a6cd2d4ff25caa687f30ab","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 18:14:13.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"b9a64f557e595f7279ebe959811516e6b849daa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 18:35:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3199877f9f8415780c76ea41938e09d : SUCCESS in 9m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87ac225a9a23436ea13b0a82948de26a : FAILURE in 18m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/54e60efca89348dc833a6ff2cc4e1117 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5afe5a04ba4657d176edb9afa1a6caa388d10b22","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-27 10:52:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c30b0173d0072d660b7f963c46ef9e8b39c0b310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 11:21:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9513476b6e67478ba6c5045e8c94cee3 : SUCCESS in 12m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d858690b08fe477fa2e04d53b75bea28 : FAILURE in 23m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ae327b02a3c147b68282747865177bfc : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":3},{"id":"ec5d0dfce24737ec3d9c8cc8b99a6eeb15a8c90f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-27 17:43:55.000000000","message":"Patch Set 3:\n\nI\u0027ve sussed why the docs build is failing here. reno is trying to find an earlier branch than \u0027stable/v4-branch\u0027, which was been deleted and replaced by \u0027v4-branch-eol\u0027 and is doing so alphabetically [1]. This causes it to select \u0027stable/train\u0027 (t \u003c v) and search for the earliest tag on _that_ branch, \u002713.0.0.0rc1\u0027, from the \u0027stable/v4-branch\u0027 branch. That\u0027s what excludes.\n\nThe solution seems to be to use stricter regexes for the various branches and tags. I\u0027ll propose that shortly.\n\n[1] https://github.com/openstack/reno/blob/3.0.0/reno/scanner.py#L850-L852","accounts_in_message":[],"_revision_number":3},{"id":"54d3d9d7980fa9a50448999770885ac1a68f0041","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 07:12:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dc8070239161a36c19042b13a6a49bfc42d4c1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 07:37:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f48cabe3d004c5c83f09503d128b430 : SUCCESS in 10m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d898247a469d4feea601c1339d4f4fc6 : SUCCESS in 19m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6e0b0e44f6f9431b9e621ffa0c07ffb3 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":3},{"id":"14052422f81744ba4a369e7daa1166004a5c403f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-02 09:44:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8ae5829313bc8851a6713a173feefdcead9325d9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-02 10:35:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bc2225075742fb6fda621de0330af0475fc2b609","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 10:35:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e5bc6a6f12a4433bf37e1add0d889aeeea1dbd03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:06:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cebddd90d3cfa02f671b8340ef07c9caa4556a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:06:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9feb52bd5094aadb4bf7fcdea1563b1 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da6bc72539664833879a200707f55b12 : SUCCESS in 12m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4b610c7b6c24961aa6adf8132a1f8da : SUCCESS in 23m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"16c12d14e732d5c554e344c5e94e3d6b3e0c1537","revisions":{"685343ca0bbe71da95c3c29a84489771fe5dadda":{"kind":"REWORK","_number":1,"created":"2020-03-26 17:44:58.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/82/715282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/82/715282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/82/715282/1"}}},"commit":{"parents":[{"commit":"97860919df69812e2a90792924584be5e3a9a031","subject":"Merge \"Use governance_consistency.py for MembershipFreeze\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97860919df69812e2a90792924584be5e3a9a031"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-03-26 17:40:37.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-03-26 17:40:37.000000000","tz":480},"subject":"Release new rocky version for heat repos","message":"Release new rocky version for heat repos\n\nChange-Id: I899b5cb74e2902c8e5d37a244ebefaa98dbf30e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/685343ca0bbe71da95c3c29a84489771fe5dadda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/685343ca0bbe71da95c3c29a84489771fe5dadda"}]},"branch":"refs/heads/master"},"09fa821743aa32ba41f7936c6d1280c9bd805e1d":{"kind":"REWORK","_number":2,"created":"2020-03-26 18:12:32.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/82/715282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/82/715282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/82/715282/2"}}},"commit":{"parents":[{"commit":"97860919df69812e2a90792924584be5e3a9a031","subject":"Merge \"Use governance_consistency.py for MembershipFreeze\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97860919df69812e2a90792924584be5e3a9a031"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-03-26 17:40:37.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-03-26 18:12:17.000000000","tz":480},"subject":"Release new rocky version for heat repos","message":"Release new rocky version for heat repos\n\nChange-Id: I899b5cb74e2902c8e5d37a244ebefaa98dbf30e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/09fa821743aa32ba41f7936c6d1280c9bd805e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/09fa821743aa32ba41f7936c6d1280c9bd805e1d"}]},"branch":"refs/heads/master"},"16c12d14e732d5c554e344c5e94e3d6b3e0c1537":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-27 10:52:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/715282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/82/715282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/82/715282/3"}}},"commit":{"parents":[{"commit":"fb8a2ec5d89d2daa7c317a2282d121d15bcdadc2","subject":"Merge \"Release 3.1.0 of cliff\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb8a2ec5d89d2daa7c317a2282d121d15bcdadc2"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-03-26 17:40:37.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-03-27 10:52:16.000000000","tz":0},"subject":"Release new rocky version for heat repos","message":"Release new rocky version for heat repos\n\nChange-Id: I899b5cb74e2902c8e5d37a244ebefaa98dbf30e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16c12d14e732d5c554e344c5e94e3d6b3e0c1537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16c12d14e732d5c554e344c5e94e3d6b3e0c1537"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
