)]}'
{"id":"openstack%2Freleases~715382","triplet_id":"openstack%2Freleases~master~I1ebf82909cb6c75519a7e1a004ae67a88dfb17f1","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I1ebf82909cb6c75519a7e1a004ae67a88dfb17f1","subject":"Team OpenStackClient is merged into OpenStackSDK","status":"MERGED","created":"2020-03-27 10:22:09.000000000","updated":"2020-03-27 11:19:31.000000000","submitted":"2020-03-27 11:19:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715382-1585307971457-795c03e7","meta_rev_id":"ec6d5913897c9f62bb125855289c4e38b345f5c4","_number":715382,"virtual_id_number":715382,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-27 11:19:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-03-27 11:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-03-27 11:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 11:06:07.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-03-27 11:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5715f85e07c14c70b01ba6d776fb53ab258540c9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-27 10:22:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c58941b488631bf520de660b47deac7caadb1b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 10:35:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f12f783d581d488b9c56e3ceda60d926 : SUCCESS in 11m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb87d407c61a4c36a67c5eb7ba90d9be : SUCCESS in 7m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ab34b86a43f34ebab78ec7d191340ccd : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fd73f7c6d33616093c3bdc1663818fb946170079","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-27 11:06:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerging it ASAP will avoid other PTL-Approved false negatives","accounts_in_message":[],"_revision_number":1},{"id":"1ee33985ea6aac39ebae6bf07a8f3664f3c58734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 11:06:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"678431ffc0f63dfa7b8824febb3b37930561cee8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 11:19:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec6d5913897c9f62bb125855289c4e38b345f5c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 11:19:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db310cc058c14ba7a1c86f4eb1e815c2 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00448cf175bf4ee695ba30149d646c0d : SUCCESS in 12m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2492040bf28940a6954725050caadb89 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebd018689940e08bf067039869285e72ba9b7234","revisions":{"ebd018689940e08bf067039869285e72ba9b7234":{"kind":"REWORK","_number":1,"created":"2020-03-27 10:22:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/82/715382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/82/715382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/715382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/82/715382/1"}}},"commit":{"parents":[{"commit":"632adf387b2b665968e90c6c1083bea5220ec959","subject":"Release 3.1.0 of cliff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/632adf387b2b665968e90c6c1083bea5220ec959"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 10:18:47.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 10:21:42.000000000","tz":60},"subject":"Team OpenStackClient is merged into OpenStackSDK","message":"Team OpenStackClient is merged into OpenStackSDK\n\nChange Ib1088af1e56e083ee20e8b1623ec7d0fb4335334 merged the\nOpenStackClient and OpenStackSDK teams into OpenStackSDK, follow this\nfor all deliverables.\n\nChange-Id: I1ebf82909cb6c75519a7e1a004ae67a88dfb17f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ebd018689940e08bf067039869285e72ba9b7234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ebd018689940e08bf067039869285e72ba9b7234"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
