)]}'
{"id":"openstack%2Freleases~716069","triplet_id":"openstack%2Freleases~master~I9ad3b891c09dab47e38de4d77dcea6d3b1516bf8","project":"openstack/releases","branch":"master","topic":"oslo-master","hashtags":[],"change_id":"I9ad3b891c09dab47e38de4d77dcea6d3b1516bf8","subject":"Oslo releases for 2020-03-30","status":"MERGED","created":"2020-03-30 19:59:39.000000000","updated":"2020-03-31 14:25:39.000000000","submitted":"2020-03-31 14:25:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716069-1585664739984-cfc8ac43","meta_rev_id":"9e1d4079611786a3a07e5f870d718b373c2c931e","_number":716069,"virtual_id_number":716069,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 14:25:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-31 14:08:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-31 09:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-31 14:08:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-30 20:01:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 09:36:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-31 14:08:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-31 14:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3167e61a9364a0a0ea0075550519a3544ceead7","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-03-30 19:59:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c54999e182cd9b2f614ae9e993731702d40623f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 20:01:43.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"43602cc998fcef0e429ebca93046262e6addcc77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 20:15:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2492cb5030734ca1b07df5cd371e8f37 : SUCCESS in 10m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55bfc170c5934e199444b8072ba18814 : SUCCESS in 8m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/65fb05a0ba6f41b8863e044a5ce9d2e5 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"dc17c4bf7b6888cfa32d16818cd155569081204e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-31 09:36:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"38331f25f74d8dd07760a96874434b94b50ac6ab","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-31 14:08:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nVersion bumps make sense, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"ad076b1ebb66a3d3ebb53f0529e60f81dd913255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:09:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eeeec1b70a1b43d00bee6be4cc7180a5d0482c39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:25:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b65ed30ee79afa32a3756e76f962706b4f8aac99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:25:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d3e256e855743689186ebd2b1abfa90 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd076479acb64fe18d9bf229cc780046 : SUCCESS in 13m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce1c84f132f142848ffa020cbf81b0c1 : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e3ec8ca9278d01aa2677e20d67958ea2f3ac2a1","revisions":{"7e3ec8ca9278d01aa2677e20d67958ea2f3ac2a1":{"kind":"REWORK","_number":1,"created":"2020-03-30 19:59:39.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/69/716069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/716069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/716069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/716069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/716069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/716069/1"}}},"commit":{"parents":[{"commit":"7b4aac601cdd3c8d147fd9e029a42d8a756c89ce","subject":"Merge \"Release OSA Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b4aac601cdd3c8d147fd9e029a42d8a756c89ce"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-03-30 19:57:56.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-03-30 19:57:56.000000000","tz":0},"subject":"Oslo releases for 2020-03-30","message":"Oslo releases for 2020-03-30\n\nMajor for castellan because one of the changes removes a public member\nfrom some of their modules. Feature for oslo.vmware because I think\nthe change has an implied dependency bump on the Glance client, even\nthough the client is passed in and not explicitly listed in the\nrequirements file.\n\nChange-Id: I9ad3b891c09dab47e38de4d77dcea6d3b1516bf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e3ec8ca9278d01aa2677e20d67958ea2f3ac2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e3ec8ca9278d01aa2677e20d67958ea2f3ac2a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
