)]}'
{"id":"openstack%2Freleases~716755","triplet_id":"openstack%2Freleases~master~Ie486e6a25460b04a54f6f69e3afe74876bc31501","project":"openstack/releases","branch":"master","topic":"u-final-lib","hashtags":[],"change_id":"Ie486e6a25460b04a54f6f69e3afe74876bc31501","subject":"Release final mistral-extra for ussuri","status":"MERGED","created":"2020-04-01 20:10:55.000000000","updated":"2020-04-04 16:33:16.000000000","submitted":"2020-04-04 16:33:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716755-1586017996126-dc9227fa","meta_rev_id":"489baeedf2926a500234f189ccdaeebe44c8fc64","_number":716755,"virtual_id_number":716755,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 16:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-04 16:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-03 14:48:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-04-03 02:48:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-04 16:17:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-03 02:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 02:48:31.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-03 14:48:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-04 16:33:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f795df4226f8e8a3e6aade7f74a9cf3a53001a7d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-01 20:10:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a90e89c31b4ff7b75416e1ded855efb18782ec3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-01 20:24:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"875ae31bd9545dc69f2f2437e1cc598b4ede68c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 20:39:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f684cedb31cf4c93b3bf557aaeb8bf7f : SUCCESS in 10m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c48a374b81154653972a489ce89e3671 : SUCCESS in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1f99e9fc792a4ff39fd71cfd644c8ac5 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0c5f077f701bab35577a09a9de911684c0579162","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-02 09:59:10.000000000","message":"Patch Set 2:\n\nAre sure we can consider [1] as a fix?\n\nIt rather looks like to a new feature (new entrypoint etc...).\n\nThoughts?\n\n[1] https://github.com/openstack/mistral-extra/commit/b615bd4b48a97e8f90dc64cd89e890cc2e688a43","accounts_in_message":[],"_revision_number":2},{"id":"57087525005f7662b8d7ccdcb9268b4f39ccef32","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-03 02:48:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"599a3557724b676cd22120e0232853efcda66fbc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-03 02:49:31.000000000","message":"Patch Set 2:\n\nAre you planning to do the same for mistral-lib?","accounts_in_message":[],"_revision_number":2},{"id":"1a776f322880c6c380f771edbe20e0ddaeed4858","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 02:49:41.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3e388b7c5415b4587c7ab080068269f6caf82ae6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-03 14:48:00.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003e Are you planning to do the same for mistral-lib?\n\nI think we could say yes but I let\u0027s Sean confirm that.\nIf you want to cut on a specific change then you could do this too on your side (for mistral-lib).\n\nAny update about my previous question and my concern about the new entrypoint and the fact that we released a fixing version instead of a minore version?\n\nOnly +1 for now by waiting more updates","accounts_in_message":[],"_revision_number":2},{"id":"c95747b5b01da7e60c0c934aa631ff55a7fb737f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-04 16:17:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n\u003e Are you planning to do the same for mistral-lib?\n\nThere weren\u0027t any changes merged to mistral-lib when I checked. If something has merged since April 1 that needs to be released, please propose a release patch.","accounts_in_message":[],"_revision_number":2},{"id":"54c38c980913f8f4e3aef9487e0633ddad39f416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 16:17:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ba356670a059f27eedb20ce193b1f6a15791ada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 16:33:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331dc091474e436badb86607bd3cb9c2 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a750b7abced4a269f97206c740c58a6 : SUCCESS in 12m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0f331ae7a3f4897978bbd85ffe5c683 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a1dcddeb4ca6cd206c4bd152c3cfae8c85cbdda2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 16:33:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a333fec7cc76402bee847c31e2ab9e70a48857c4","revisions":{"0edb34eb757e639986e9a92938c1326374557d4c":{"kind":"REWORK","_number":1,"created":"2020-04-01 20:10:55.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/55/716755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/716755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/716755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/716755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/716755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/716755/1"}}},"commit":{"parents":[{"commit":"ff23dfdb4222b90295f85f3d8bbfd73c4ad04991","subject":"Release glance_store 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff23dfdb4222b90295f85f3d8bbfd73c4ad04991"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-01 20:10:51.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-01 20:10:51.000000000","tz":-300},"subject":"Release final mistral-extra for ussuri","message":"Release final mistral-extra for ussuri\n\nFinal ussuri non-client lib release for the April 2 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ie486e6a25460b04a54f6f69e3afe74876bc31501\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0edb34eb757e639986e9a92938c1326374557d4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0edb34eb757e639986e9a92938c1326374557d4c"}]},"branch":"refs/heads/master"},"a333fec7cc76402bee847c31e2ab9e70a48857c4":{"kind":"REWORK","_number":2,"created":"2020-04-01 20:24:05.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/55/716755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/716755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/716755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/716755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/716755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/716755/2"}}},"commit":{"parents":[{"commit":"ff23dfdb4222b90295f85f3d8bbfd73c4ad04991","subject":"Release glance_store 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff23dfdb4222b90295f85f3d8bbfd73c4ad04991"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-01 20:10:51.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-01 20:24:00.000000000","tz":-300},"subject":"Release final mistral-extra for ussuri","message":"Release final mistral-extra for ussuri\n\nFinal ussuri non-client lib release for the April 2 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ie486e6a25460b04a54f6f69e3afe74876bc31501\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a333fec7cc76402bee847c31e2ab9e70a48857c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a333fec7cc76402bee847c31e2ab9e70a48857c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
