)]}'
{"id":"openstack%2Freleases~716760","triplet_id":"openstack%2Freleases~master~I3d4678bdb3b02a70e2c136c7f7fbc4513a085bff","project":"openstack/releases","branch":"master","topic":"u-final-lib","hashtags":[],"change_id":"I3d4678bdb3b02a70e2c136c7f7fbc4513a085bff","subject":"Release final os-client-config for ussuri","status":"MERGED","created":"2020-04-01 20:15:05.000000000","updated":"2020-04-02 18:05:56.000000000","submitted":"2020-04-02 18:05:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716760-1585850756716-afcf8495","meta_rev_id":"f83b3e6bb4bcd5b3eda64b23b4870442c83a1f74","_number":716760,"virtual_id_number":716760,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 18:05:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-02 17:04:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-02 16:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-04-02 10:00:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-02 17:04:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-02 17:39:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 10:00:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-02 16:49:51.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-02 18:05:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70cce0718db60b5ed1ef41b8970df43d7600d015","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-01 20:15:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"680e176fa632ca109f8b92009edf66301c6fd1f9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-01 20:25:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a472ac4d50339b781e26dbf493258ce9bb8bb9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 20:37:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb0f8dfbb99460da5cf1887f72be6ba : SUCCESS in 9m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/417b181ff63c43e587ab88eb596159ef : SUCCESS in 6m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c3d9d8acec1e47fda31ce22fd4126ece : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4db4338a9a74ba81be0693e831161a58878d1338","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-02 10:00:33.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM waiting for PTL validation","accounts_in_message":[],"_revision_number":2},{"id":"683df933ca301b8ad6b7b04732abee0d1fed3b5a","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-02 16:49:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5e48dbc5f413bfb7b74128208f765b9061a998e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 17:04:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bc02304f95ba7f5f5dcede54e1d0eb7d5dec9b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:14:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"23f3c76a1a884b9a42d9677bfdd2730a47a5ca60","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:39:54.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"213da7bb20785396e7e02fada96254f7358a712c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:05:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9d6c98440647bfbf4fb6790e97ce9c : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c6263a10f54fc3a0bf5220c75d7516 : SUCCESS in 13m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b26c112b7c74452ca83cfcf55d5a37b0 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a69212bff6d4b9f288a584a2df56b46ea1f71429","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:05:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69733d3fae58dfc514b34fdb273185355e74c979","revisions":{"742472589eb9113653c36027df53db9815a2b75d":{"kind":"REWORK","_number":1,"created":"2020-04-01 20:15:05.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/60/716760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/716760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/716760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/716760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/716760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/716760/1"}}},"commit":{"parents":[{"commit":"ff23dfdb4222b90295f85f3d8bbfd73c4ad04991","subject":"Release glance_store 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff23dfdb4222b90295f85f3d8bbfd73c4ad04991"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-01 20:15:01.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-01 20:15:01.000000000","tz":-300},"subject":"Release final os-client-config for ussuri","message":"Release final os-client-config for ussuri\n\nFinal ussuri non-client lib release for the April 2 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: I3d4678bdb3b02a70e2c136c7f7fbc4513a085bff\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/742472589eb9113653c36027df53db9815a2b75d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/742472589eb9113653c36027df53db9815a2b75d"}]},"branch":"refs/heads/master"},"69733d3fae58dfc514b34fdb273185355e74c979":{"kind":"REWORK","_number":2,"created":"2020-04-01 20:25:14.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/60/716760/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/716760/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/716760/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/716760/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/716760/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/716760/2"}}},"commit":{"parents":[{"commit":"ff23dfdb4222b90295f85f3d8bbfd73c4ad04991","subject":"Release glance_store 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff23dfdb4222b90295f85f3d8bbfd73c4ad04991"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-01 20:15:01.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-01 20:25:10.000000000","tz":-300},"subject":"Release final os-client-config for ussuri","message":"Release final os-client-config for ussuri\n\nFinal ussuri non-client lib release for the April 2 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: I3d4678bdb3b02a70e2c136c7f7fbc4513a085bff\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/69733d3fae58dfc514b34fdb273185355e74c979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/69733d3fae58dfc514b34fdb273185355e74c979"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
