)]}'
{"id":"openstack%2Freleases~717644","triplet_id":"openstack%2Freleases~master~I755a37427b99e7151cfc06f470caf3e2ca71d3c8","project":"openstack/releases","branch":"master","topic":"add-cycle-highlights-for-zaqar","hashtags":[],"change_id":"I755a37427b99e7151cfc06f470caf3e2ca71d3c8","subject":"Add cycle highlights for Zaqar","status":"MERGED","created":"2020-04-06 01:48:13.000000000","updated":"2020-04-06 11:21:48.000000000","submitted":"2020-04-06 11:21:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717644-1586172108224-38ce0ed0","meta_rev_id":"60ab9ceb858693da0ec39250c5e24ccf7a11f9dc","_number":717644,"virtual_id_number":717644,"owner":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 11:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-04-06 09:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-04-06 11:10:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-04-06 11:10:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-06 02:24:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-06 09:28:52.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-06 11:10:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-06 11:21:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a6c319f36221403851e4b5d1d805d7a800e4eb1","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2020-04-06 01:48:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a5a20033e7e3916fe639972e0b84aa1a3de3cec","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 01:49:12.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"10c3a6712c9f3e594fa2882604551e96bda9c696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 01:58:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e56d6d8153422ea540b89113037ee8 : SUCCESS in 10m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84b619f1d3f54cc7afa638082ee81096 : FAILURE in 6m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7dd4019050c54140bca9590d9acfea07 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ce70e9ab25562873fd5b8097f2cad677b29e7e0c","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2020-04-06 02:22:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7c350897f5444ad80ca60aff85db985f383fdab","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 02:24:03.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"013ed1a81f9f2e20389a30f12876993d709afa1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 02:33:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d9f3abe2a34163ae629aad072393f1 : SUCCESS in 9m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a665558c115848c2a00d38315270ed02 : SUCCESS in 4m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c2f87ff754994e4bbddd531422ccc3a9 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"70ea343b3222d30ae37538931d13bd15529eaf34","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-06 09:28:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c606e0baed2ae216e492d054e529d39ca6e4d1d4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-06 11:10:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e1ab528ae0768fc68b0d48f53d4e889f5014e1f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 11:10:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1eab55c5b882d47c1fbf87e1c30303a2f3e1a719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 11:21:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759631b135a64105bd424794291985c6 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847da8ef5b3245d79f31fae83693f358 : SUCCESS in 10m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33ce480c07a74411a4e150e8b9b26993 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":2},{"id":"80b0c88e85f69bd581b9049122b74e3bc477c5cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 11:21:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1d2b819c5e9332efad0ec16473557e32d9d36ee","revisions":{"f7f0916d505022799692b3f0f8a9ba9df50bf71c":{"kind":"REWORK","_number":1,"created":"2020-04-06 01:48:13.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/44/717644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/717644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/717644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/717644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/717644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/717644/1"}}},"commit":{"parents":[{"commit":"4d738658b9b580717dd6442f95d83cbc43b68bcb","subject":"Merge \"Release final monasca-common for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d738658b9b580717dd6442f95d83cbc43b68bcb"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-04-06 01:43:18.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-04-06 01:43:18.000000000","tz":480},"subject":"Add cycle highlights for Zaqar","message":"Add cycle highlights for Zaqar\n\nChange-Id: I755a37427b99e7151cfc06f470caf3e2ca71d3c8\nSigned-off-by: wanghao \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7f0916d505022799692b3f0f8a9ba9df50bf71c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7f0916d505022799692b3f0f8a9ba9df50bf71c"}]},"branch":"refs/heads/master"},"c1d2b819c5e9332efad0ec16473557e32d9d36ee":{"kind":"REWORK","_number":2,"created":"2020-04-06 02:22:56.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/44/717644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/717644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/717644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/717644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/717644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/717644/2"}}},"commit":{"parents":[{"commit":"4d738658b9b580717dd6442f95d83cbc43b68bcb","subject":"Merge \"Release final monasca-common for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d738658b9b580717dd6442f95d83cbc43b68bcb"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-04-06 01:43:18.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2020-04-06 02:21:54.000000000","tz":480},"subject":"Add cycle highlights for Zaqar","message":"Add cycle highlights for Zaqar\n\nChange-Id: I755a37427b99e7151cfc06f470caf3e2ca71d3c8\nSigned-off-by: wanghao \u003csxmatch1986@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1d2b819c5e9332efad0ec16473557e32d9d36ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1d2b819c5e9332efad0ec16473557e32d9d36ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
