)]}'
{"id":"openstack%2Freleases~717939","triplet_id":"openstack%2Freleases~master~I85f69c1ed99d8b8c0e51ad4be56821e914b25ab2","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I85f69c1ed99d8b8c0e51ad4be56821e914b25ab2","subject":"Release python-zunclient 4.0.0","status":"MERGED","created":"2020-04-07 03:06:42.000000000","updated":"2020-04-07 20:32:56.000000000","submitted":"2020-04-07 20:32:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717939-1586291576797-35f7c261","meta_rev_id":"c50b4b906a6afaf4b4d36551620b067731465c14","_number":717939,"virtual_id_number":717939,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-07 13:30:28.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 20:32:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-04-07 20:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-07 13:51:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-04-07 20:13:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-07 13:28:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 13:51:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-07 20:13:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-07 20:32:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"786e3de1e966488d5840af23dc73727c8c64dcf0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-07 03:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21888d6cddc698ce019db59cb8efb2a8fced8990","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 03:07:49.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"3769eb08673b74622fcd742d11f31939e2b2de1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 03:20:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ca6a6025f74494b8e7691def6e1e27 : SUCCESS in 9m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4d48476df1e4ed88ee29595ccd05081 : SUCCESS in 4m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e3277c198a4e4f36b76bbf3544ebf75c : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"19fd582f0e7b0947ddddd9fec00ba3d862c82fc7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-07 09:35:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nHello\n\nSee my inline comment","accounts_in_message":[],"_revision_number":1},{"id":"3ee7455d4c1471b6e446e0556d295f8fd9b98db6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-07 13:26:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92c225d585724228b0d28ba1d6d4bb36ef391ce3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:28:47.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"e7f3011eab07c828407754c6f0659e80fc2e7b7a","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-04-07 13:30:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2fedb9412fcef604a462d32b3202218218b9e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:42:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d310ec41da90490daa44f9aaaffb4474 : SUCCESS in 13m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1a74591cb644e06b8f4bda4bb946ef6 : SUCCESS in 8m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0736cb73413c4d34a9a1a0f13c539660 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"6ae35a666be504d5bcb3f9154e01b94821de6c7c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-07 13:51:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4da4b8fcea74ee5a3b46ce222309e8554291b209","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 20:13:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"86717784bd10857d28607e272d6ab18933bc550c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:13:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf2d9bf887dc83c20355ffc5e976f8d4351573e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:32:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c50b4b906a6afaf4b4d36551620b067731465c14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:32:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c3f8c20bdf54f26a53096ccb9bbe6a1 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0779b5ac2e048d491ad159f1cc8c439 : SUCCESS in 11m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58c8392a3c964e438176ed2c9bd23b2b : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"01a0ce337b7acedac41ef53d289db1f6f7b59057","revisions":{"11ccd77ab608f08baa481b61122baf0ab616b6ed":{"kind":"REWORK","_number":1,"created":"2020-04-07 03:06:42.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/39/717939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/39/717939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/717939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/717939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/717939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/39/717939/1"}}},"commit":{"parents":[{"commit":"ff7dc75a27be569fa095e394fb0f582dabf96580","subject":"Merge \"Add ussuri cycle-highlights for manila\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff7dc75a27be569fa095e394fb0f582dabf96580"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-04-07 03:06:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-04-07 03:06:20.000000000","tz":0},"subject":"Release python-zunclient 3.7.0","message":"Release python-zunclient 3.7.0\n\nChange-Id: I85f69c1ed99d8b8c0e51ad4be56821e914b25ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11ccd77ab608f08baa481b61122baf0ab616b6ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11ccd77ab608f08baa481b61122baf0ab616b6ed"}]},"branch":"refs/heads/master"},"01a0ce337b7acedac41ef53d289db1f6f7b59057":{"kind":"REWORK","_number":2,"created":"2020-04-07 13:26:24.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/39/717939/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/39/717939/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/717939/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/717939/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/717939/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/39/717939/2"}}},"commit":{"parents":[{"commit":"ff7dc75a27be569fa095e394fb0f582dabf96580","subject":"Merge \"Add ussuri cycle-highlights for manila\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ff7dc75a27be569fa095e394fb0f582dabf96580"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-04-07 03:06:20.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-04-07 13:26:11.000000000","tz":0},"subject":"Release python-zunclient 4.0.0","message":"Release python-zunclient 4.0.0\n\nChange-Id: I85f69c1ed99d8b8c0e51ad4be56821e914b25ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01a0ce337b7acedac41ef53d289db1f6f7b59057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01a0ce337b7acedac41ef53d289db1f6f7b59057"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
