)]}'
{"id":"openstack%2Freleases~718144","triplet_id":"openstack%2Freleases~master~Ie7a8f36c3b88f5451c44c5a37f1ac3543a942e6e","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ie7a8f36c3b88f5451c44c5a37f1ac3543a942e6e","subject":"Create stable/1.0 branch for Tenks 1.0.x","status":"MERGED","created":"2020-04-07 15:10:02.000000000","updated":"2020-04-07 20:32:53.000000000","submitted":"2020-04-07 20:32:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718144-1586291573304-e3c9830e","meta_rev_id":"839417c8432fdeb74f3f18d63aa75ee62a76e43f","_number":718144,"virtual_id_number":718144,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 20:32:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-07 20:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-07 20:06:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-07 17:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-07 20:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-07 17:13:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 17:22:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-07 20:12:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-07 20:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abeea4a35da0185956a7db1a8cce4f57bb1ed029","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 15:10:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa518f2740c3debee77e72cd37616a1192a48d52","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-07 15:11:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"52c05751941d107eaca5cd386cca1e6b487d3e2e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:13:25.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"b0ea690269dfe02d31f20aa3b5119a447584cb5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:27:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680036c2c36843768c1c9ca9994bbeb7 : SUCCESS in 13m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/240ac9e861534ffa9eaada67307a035c : FAILURE in 7m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f5cb36bfb6574386994cd91e92cf85f4 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"781551ea3a9e625560adb541d52a68e56673d2fb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-07 15:29:11.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"862675aa153fb2f5e1f26ed082470dff65b29df8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 15:33:48.000000000","message":"Patch Set 1:\n\ndeliverables/_independent/tenks.yaml: validate_stable_branches: stable branches must be named for known series but stable/1.0 was not found in [\u0027austin\u0027, \u0027bexar\u0027, \u0027cactus\u0027, \u0027diablo\u0027, \u0027essex\u0027, \u0027folsom\u0027, \u0027grizzly\u0027, \u0027havana\u0027, \u0027icehouse\u0027, \u0027juno\u0027, \u0027kilo\u0027, \u0027liberty\u0027, \u0027mitaka\u0027, \u0027newton\u0027, \u0027ocata\u0027, \u0027pike\u0027, \u0027queens\u0027, \u0027rocky\u0027, \u0027stein\u0027, \u0027train\u0027, \u0027ussuri\u0027]\n\nIs that a valid check for independently released projects?","accounts_in_message":[],"_revision_number":1},{"id":"aba2c9bd1c094a16680fd524ec32a736e95d7099","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 15:34:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eb6358268579e19a66738738e2a19eeb88137018","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:37:12.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"fcdc767fd6bfdfab83fc7d7509be6f8f89f8a46d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 16:01:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3a132de01d4e2598f6d17e877b90d9 : SUCCESS in 12m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12097db394fc4b4a9d13a831ad6e99ff : FAILURE in 7m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ee5dbe103953432bb837d5e5be05bbc2 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b039e01e601b7b5626144579143907170e96dea0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 16:23:54.000000000","message":"Patch Set 2:\n\n:(\n\ndeliverables/_independent/tenks.yaml: validate_branch_prefixes: branch name stable-1.0 does not use a valid prefix: {\u0027feature\u0027, \u0027stable\u0027}","accounts_in_message":[],"_revision_number":2},{"id":"c7d550ae9691f9fb672b51938c984708cecce619","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 17:08:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8edd97dd4ceb1b10c616f38d45e6d042d2ca9b8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:13:32.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"15a92a0e9658edfe188d1395c232e7d942575f37","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-07 17:22:33.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks Sean!","accounts_in_message":[],"_revision_number":3},{"id":"d92251635d70aa20f82a9747d492071d89772a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:24:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd95f6689334e4db0f23d1b87ccee5a : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50d820576a7f4f608baaf96979fdb706 : SUCCESS in 9m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0d80420852d42a7b1e5d5bdc8249be7 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a44dd4c7cfa74f06bfc0c1e895bafdb3 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c66ef10d5554fe8926ef491dbb72d68 : SUCCESS in 12m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b87af9f774384efbb58e129fb39a32b0 : SUCCESS in 7m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e2494f65c8154ee59dda518c2c41f2df : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":3},{"id":"6091c6ca8f3d7de7a72dbc61bfaded0965cb30c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 20:06:48.000000000","message":"Patch Set 3: Code-Review+1\n\nPerfect, thanks!","accounts_in_message":[],"_revision_number":3},{"id":"cd99cf4ded459b9b602223dd5c5534e50dcdea9b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 20:12:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f925e8e1002f6e31844b72f436ecb1dfc8e228a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:12:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aa551588e6f081e42f17f3acbd4a9caf2597836c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:32:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"839417c8432fdeb74f3f18d63aa75ee62a76e43f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:32:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624de5c872204769bb3469136cd2019b : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3f4f7f2ccd8479899b89ff293e8a69f : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1af377047e5478b833e3eac7329e995 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730f9df2d88b4addbc67e99157aadccf : SUCCESS in 12m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcc87f838e394a218e9a3a7d7172617a : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"21b5834257f72d7577053560229b047b13b22b07","revisions":{"f1912a0b1465b5b5b714b1b716fa9dd3371338e1":{"kind":"REWORK","_number":1,"created":"2020-04-07 15:10:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/718144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/718144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/718144/1"}}},"commit":{"parents":[{"commit":"9708199e777788783a54fb03888a42a2889f6af7","subject":"Release python-glanceclient 3.1.0 for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9708199e777788783a54fb03888a42a2889f6af7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-07 15:08:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-07 15:09:07.000000000","tz":60},"subject":"Create stable/1.0 branch for Tenks 1.0.x","message":"Create stable/1.0 branch for Tenks 1.0.x\n\nThis will allow us to create releases with Python 2 support, while\nmoving forward with the Python 2 drop on master.\n\nChange-Id: Ie7a8f36c3b88f5451c44c5a37f1ac3543a942e6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1912a0b1465b5b5b714b1b716fa9dd3371338e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1912a0b1465b5b5b714b1b716fa9dd3371338e1"}]},"branch":"refs/heads/master"},"faaf3a30b049e8509b97d415f44c4e4943f026a3":{"kind":"REWORK","_number":2,"created":"2020-04-07 15:34:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/718144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/718144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/718144/2"}}},"commit":{"parents":[{"commit":"9708199e777788783a54fb03888a42a2889f6af7","subject":"Release python-glanceclient 3.1.0 for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9708199e777788783a54fb03888a42a2889f6af7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-07 15:08:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-07 15:33:58.000000000","tz":0},"subject":"Create stable/1.0 branch for Tenks 1.0.x","message":"Create stable/1.0 branch for Tenks 1.0.x\n\nThis will allow us to create releases with Python 2 support, while\nmoving forward with the Python 2 drop on master.\n\nChange-Id: Ie7a8f36c3b88f5451c44c5a37f1ac3543a942e6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faaf3a30b049e8509b97d415f44c4e4943f026a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faaf3a30b049e8509b97d415f44c4e4943f026a3"}]},"branch":"refs/heads/master"},"21b5834257f72d7577053560229b047b13b22b07":{"kind":"REWORK","_number":3,"created":"2020-04-07 17:08:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/44/718144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/718144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/718144/3"}}},"commit":{"parents":[{"commit":"9708199e777788783a54fb03888a42a2889f6af7","subject":"Release python-glanceclient 3.1.0 for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9708199e777788783a54fb03888a42a2889f6af7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-07 15:08:51.000000000","tz":60},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 17:08:28.000000000","tz":-300},"subject":"Create stable/1.0 branch for Tenks 1.0.x","message":"Create stable/1.0 branch for Tenks 1.0.x\n\nThis will allow us to create releases with Python 2 support, while\nmoving forward with the Python 2 drop on master.\n\nChange-Id: Ie7a8f36c3b88f5451c44c5a37f1ac3543a942e6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21b5834257f72d7577053560229b047b13b22b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21b5834257f72d7577053560229b047b13b22b07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
