)]}'
{"id":"openstack%2Freleases~718244","triplet_id":"openstack%2Freleases~master~Ia8a3fdcce4b4865d7ba787f5fe322e995efc8a75","project":"openstack/releases","branch":"master","topic":"ussuri-client-lib","hashtags":[],"change_id":"Ia8a3fdcce4b4865d7ba787f5fe322e995efc8a75","subject":"Release final python-congressclient for ussuri","status":"MERGED","created":"2020-04-07 22:31:58.000000000","updated":"2020-04-09 10:44:36.000000000","submitted":"2020-04-09 10:44:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718244-1586429076476-a17bf472","meta_rev_id":"f9d01c84d85e52624468e7ab31051b5ebabbe819","_number":718244,"virtual_id_number":718244,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 10:44:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-04-09 10:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-08 23:59:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-04-09 10:25:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-09 00:01:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 23:59:16.000000000","updated_by":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"reviewer":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"state":"REVIEWER"},{"updated":"2020-04-09 10:25:12.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-09 10:44:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9706fa18d3f4351aa8d74c092c28e278691fec6a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 22:31:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdb6973a153813a0d0164b8e53196f26194136b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:08:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f600e019c2ef47a69c4c01ede5f522dd : SUCCESS in 14m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8768eb8aaf74824b3e19200562d0144 : SUCCESS in 8m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/64b5f7b8c12a4f58a34bf00702174899 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c22095d973f92e3f04ff08971d980fd5fafe2093","author":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"date":"2020-04-08 23:59:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cacbe9027c959a5856bbd13f72b7aab1100396fa","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 00:01:35.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"96a92797c0a2232382d71706f15ae775c47de19c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-09 10:25:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd9b62d625deb686fff4fe46a3a7e1e874915e78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:25:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32f29770204dae76cf10abada7edf608d4cef1fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:44:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b111325bfaf41839b372f188bd870db : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a071229641bb478d945957da2cd302cc : SUCCESS in 12m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c80ddbe555641bbafa41b16087b4e87 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"de84c961809f301c140f006563d9d1989c8f451a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:44:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5d3d30a00974345580edee91b37d786df704fdf","revisions":{"e5d3d30a00974345580edee91b37d786df704fdf":{"kind":"REWORK","_number":1,"created":"2020-04-07 22:31:58.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/44/718244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/718244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/718244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/718244/1"}}},"commit":{"parents":[{"commit":"cd46003d839f520e0310494750c68c9d85596797","subject":"Merge \"Release python-zunclient 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd46003d839f520e0310494750c68c9d85596797"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:31:56.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:31:56.000000000","tz":-300},"subject":"Release final python-congressclient for ussuri","message":"Release final python-congressclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ia8a3fdcce4b4865d7ba787f5fe322e995efc8a75\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e5d3d30a00974345580edee91b37d786df704fdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e5d3d30a00974345580edee91b37d786df704fdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
