)]}'
{"id":"openstack%2Freleases~718260","triplet_id":"openstack%2Freleases~master~Ib6dfb66f318129b11e10ab5084ccd9c6a5291856","project":"openstack/releases","branch":"master","topic":"ussuri-client-lib","hashtags":[],"change_id":"Ib6dfb66f318129b11e10ab5084ccd9c6a5291856","subject":"Release final python-novaclient for ussuri","status":"MERGED","created":"2020-04-07 22:38:13.000000000","updated":"2020-04-11 10:44:37.000000000","submitted":"2020-04-11 10:44:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718260-1586601877466-a663f06a","meta_rev_id":"b39d561e049a150c8321fcffe894443596500a49","_number":718260,"virtual_id_number":718260,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-11 10:44:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-11 10:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2020-04-11 07:50:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-11 10:32:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-11 07:52:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 08:57:52.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-04-11 02:56:02.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2020-04-11 07:50:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-04-11 10:44:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53ff2b1e098be1f85389bfe075a473bff4b3e331","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 22:38:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f747c05a06969eba0d4f0ea917ed72a0aa15277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:09:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d377de17b144719898c1f8f6a5e2d5 : SUCCESS in 14m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0f0a2081e9f4685969830a18af375af : SUCCESS in 8m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3302f5a25c2249e88f4fb3fb1284b0ee : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f77b36b257645f20aeddaa13974cb2305db169ba","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-08 12:44:59.000000000","message":"Patch Set 1: Code-Review-1\n\nWaiting for [1] in nova and then [2] in novaclient to land\n\n[1] https://review.opendev.org/#/c/701430/\n[2] https://review.opendev.org/#/c/714956/","accounts_in_message":[],"_revision_number":1},{"id":"0eca33412fad0f5a0dcf6f5bf753919a169337cd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-09 11:30:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"62185b353780553c38386a877f20c0459694a4e1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-09 11:30:28.000000000","message":"Patch Set 2: Code-Review-1\n\nWaiting for [1] in nova and then [2] in novaclient to land\n\n[1] https://review.opendev.org/#/c/701430/\n[2] https://review.opendev.org/#/c/714956/","accounts_in_message":[],"_revision_number":2},{"id":"8d8afcd4994e2a3b99d1421144d41f5e9718a35d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 11:47:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c20c2a916146b78eb997c03519511c : SUCCESS in 12m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a88f79156ab43cba4697776bf8fceb9 : FAILURE in 6m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cf46741621f84058a08d6f9ade1fac2d : FAILURE in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a79007709bbc9bb31aa05185bc65afb467894a07","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 01:35:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"88de6ec673879175c4b04dc1c2774bd3eea0e082","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 01:35:59.000000000","message":"Patch Set 3: Code-Review-1\n\nWe can wait a little longer. Just adding stable branching.","accounts_in_message":[],"_revision_number":3},{"id":"13a4602f52590a3a92dd7fc50b3386fde78d73bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 01:48:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5749ff9f12eb4059b410694fe4063f28 : SUCCESS in 11m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1595bac3daa84384bfd45f08c806cc73 : FAILURE in 6m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/039c448a69204a4497c19c4b063a8a13 : FAILURE in 6m 58s","accounts_in_message":[],"_revision_number":3},{"id":"7201b64cf9d68dfd88960dc87d4e1511bcad63f9","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2020-04-10 08:57:52.000000000","message":"Patch Set 3:\n\nWe\u0027re still waiting for Zuul to merge https://review.opendev.org/#/c/714956","accounts_in_message":[],"_revision_number":3},{"id":"0cb7f75a73af2d0813e2e07592fc7552120bfc56","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2020-04-11 02:56:02.000000000","message":"Patch Set 3:\n\nhttps://review.opendev.org/#/c/714956 has already been merged.","accounts_in_message":[],"_revision_number":3},{"id":"c1ef221a2499d116b7cf7a58a59b164d033ae708","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2020-04-11 02:56:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"855d7624708e301f69433a1306b302f9646c825d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 03:05:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435a103c3e4d4a23bd8e31e9a55e064c : SUCCESS in 7m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d59a3e7575334876aebd9f3d714f364f : SUCCESS in 5m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1593c43a1b5a4b8fafa372d3efb53106 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":4},{"id":"9336eb7ff396b1c3285785ce2600528f91770712","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-11 07:50:59.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"17cc454a6c5ce51373f2ad21a300186085af22eb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 07:52:03.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"930ae30bdf3ed4d8be8cbc5e731d9d2f19552a80","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-11 10:32:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5cfcbda9274bdf942a68ef0079c47e64ed386039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 10:32:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"96d378255b9a3332ad86ac1694fa767f19cdda35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 10:44:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4ca8423ca945bba065e7f4989021e40853562e21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 10:44:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ac8bfdba074c4f9a64d01ac508e593 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfd8b89220e48c4b6b5110fe4cdd778 : SUCCESS in 10m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/471579d5927c451380947fdbc38fcabd : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5abfe43a61ff40f7dd6fef9447c63a3a034d3b39","revisions":{"39d7845e04f1a2a5e40dfb17abe51e9bb326ae1c":{"kind":"REWORK","_number":1,"created":"2020-04-07 22:38:13.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/60/718260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/718260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/718260/1"}}},"commit":{"parents":[{"commit":"cd46003d839f520e0310494750c68c9d85596797","subject":"Merge \"Release python-zunclient 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd46003d839f520e0310494750c68c9d85596797"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:38:11.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:38:11.000000000","tz":-300},"subject":"Release final python-novaclient for ussuri","message":"Release final python-novaclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ib6dfb66f318129b11e10ab5084ccd9c6a5291856\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39d7845e04f1a2a5e40dfb17abe51e9bb326ae1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/39d7845e04f1a2a5e40dfb17abe51e9bb326ae1c"}]},"branch":"refs/heads/master"},"7ea1f430d5422211c3d2212c4e014d41b0640b52":{"kind":"REWORK","_number":2,"created":"2020-04-09 11:30:12.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/60/718260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/718260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/718260/2"}}},"commit":{"parents":[{"commit":"cd46003d839f520e0310494750c68c9d85596797","subject":"Merge \"Release python-zunclient 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd46003d839f520e0310494750c68c9d85596797"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:38:11.000000000","tz":-300},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-04-09 11:29:35.000000000","tz":0},"subject":"Release final python-novaclient for ussuri","message":"Release final python-novaclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ib6dfb66f318129b11e10ab5084ccd9c6a5291856\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ea1f430d5422211c3d2212c4e014d41b0640b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ea1f430d5422211c3d2212c4e014d41b0640b52"}]},"branch":"refs/heads/master"},"6dd5abade7f6573e8c55d87748c7feec2ea9811f":{"kind":"REWORK","_number":3,"created":"2020-04-10 01:35:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/60/718260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/718260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/718260/3"}}},"commit":{"parents":[{"commit":"cd46003d839f520e0310494750c68c9d85596797","subject":"Merge \"Release python-zunclient 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd46003d839f520e0310494750c68c9d85596797"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:38:11.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 01:35:41.000000000","tz":-300},"subject":"Release final python-novaclient for ussuri","message":"Release final python-novaclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ib6dfb66f318129b11e10ab5084ccd9c6a5291856\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dd5abade7f6573e8c55d87748c7feec2ea9811f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6dd5abade7f6573e8c55d87748c7feec2ea9811f"}]},"branch":"refs/heads/master"},"5abfe43a61ff40f7dd6fef9447c63a3a034d3b39":{"kind":"REWORK","_number":4,"created":"2020-04-11 02:56:30.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/60/718260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/718260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/718260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/718260/4"}}},"commit":{"parents":[{"commit":"aaeaf0c04aef232dd47f9b944314e5b3388b82bc","subject":"Merge \"[manila] Extend cycle highlights for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aaeaf0c04aef232dd47f9b944314e5b3388b82bc"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:38:11.000000000","tz":-300},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2020-04-11 02:54:14.000000000","tz":540},"subject":"Release final python-novaclient for ussuri","message":"Release final python-novaclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ib6dfb66f318129b11e10ab5084ccd9c6a5291856\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5abfe43a61ff40f7dd6fef9447c63a3a034d3b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5abfe43a61ff40f7dd6fef9447c63a3a034d3b39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
