)]}'
{"id":"openstack%2Freleases~718268","triplet_id":"openstack%2Freleases~master~Ibfc9469683fa9019d21c42b081681833193602cf","project":"openstack/releases","branch":"master","topic":"ussuri-client-lib","hashtags":[],"change_id":"Ibfc9469683fa9019d21c42b081681833193602cf","subject":"Release final python-tackerclient for ussuri","status":"MERGED","created":"2020-04-07 22:41:00.000000000","updated":"2020-04-10 12:18:08.000000000","submitted":"2020-04-10 12:18:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718268-1586521088151-462d5a30","meta_rev_id":"23d784e472f2fa626ab14232c146674fb3f6e1f3","_number":718268,"virtual_id_number":718268,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 12:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-10 12:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":1,"date":"2020-04-10 08:43:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-10 12:08:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 06:04:59.000000000","updated_by":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"reviewer":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"state":"REVIEWER"},{"updated":"2020-04-10 08:43:53.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2020-04-10 12:18:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5f70657e11852abdcf9a4ecec5af80de8a0276b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 22:41:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a5c7df86b43b3e08afa0aea13f275e7d715091f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:07:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6ef4ab378d44fcadd43d899aca0a4e : SUCCESS in 12m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/030360bde9af48d2852c8ceaa44d5117 : SUCCESS in 9m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/341ef3d8391645d5886d8e8a8b8a48b4 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"08c2b2f05adc83f64d7c0d9044899b4f1645208b","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-04-08 01:47:48.000000000","message":"Patch Set 1: Code-Review-1\n\nholding for now","accounts_in_message":[],"_revision_number":1},{"id":"5968e448c749efc240e35b0adee24e701d5edf4a","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-04-09 15:31:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7d5be514a0c46a52aac2e8efaa86a0173cc666d5","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-04-09 15:31:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1760b3868fcd391eeb5a3b597e7750fab735617e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:49:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d414adbfda8a4556918257d636c6ed73 : SUCCESS in 11m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff837f1ebd0c463f961b64917686060c : SUCCESS in 8m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e183c84e48d94cb7a7e44a5ac93c8c81 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5291613dcfa28b9fca199f1557b29d1d8000a927","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 01:42:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"226e63ab1ff27bca49d07b6fd53d41c6d61cf3ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 01:58:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b8ff105cda4d229662cd1a6f61ba69 : SUCCESS in 10m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22e18485f2d84257a648abadf2e2ba7d : SUCCESS in 6m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8bac401734c54df9bd9c9c22aa31e9f6 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":3},{"id":"adc009be7b19ed392d4b5c6cdd53b72fe7a607b6","author":{"_account_id":1011,"name":"Tushar Patil","email":"tushar.vitthal.patil@gmail.com","username":"tpatil"},"date":"2020-04-10 06:04:59.000000000","message":"Patch Set 3:\n\nNeed review for patch : https://review.opendev.org/#/c/716240","accounts_in_message":[],"_revision_number":3},{"id":"0e20f808186517dd96aace2199c4feeafd86b111","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-04-10 07:13:15.000000000","message":"Patch Set 3: Code-Review-1\n\n\u003e Need review for patch : https://review.opendev.org/#/c/716240\n\nHolding this patch for now. Once the above patch merged, i will update it.","accounts_in_message":[],"_revision_number":3},{"id":"69f3db3edf7ebce4b78c2b0cde9a05c9de2ffdd3","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-04-10 07:49:24.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"f14d80f134563e770b6d1fc42ced9a25cd6c775a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:01:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1003139e7f6746f1bdf18a3acdc5e713 : SUCCESS in 11m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/773b43923c9a43c880667718d263fa0b : SUCCESS in 7m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9c545f5790154fa89bf9403099f2bcfa : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4},{"id":"8dd532b94b1487b0fc05c756e8ed308a0334238b","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2020-04-10 08:43:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b393b7339323561a4f79bdbdf92a33ff5e37377f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 12:08:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nNot sure why PTL-Approved didn\u0027t show up, but have update and ack from Dharmendra.\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"b9d380723d8fa4cb461a4068febc617d1b4ef289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 12:08:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5f06c076c09948ff2b11b6bc43f5a4b3bf62c887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 12:18:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad54dcbcf614accbd5335d98fd923e8 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5ede7ffa20429cb1573246c1a6cee5 : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae85b24187e5432a8c93f1b03cdb31ed : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":4},{"id":"5d73dd32841907d66bd7499d8c9758e307a02566","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 12:18:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"27eeef91a14567966c7f8da9dbb35e15e3d3a9c2","revisions":{"cef33c45c66b89f5c3cb1cc3dd71d41b8b3dd8b8":{"kind":"REWORK","_number":1,"created":"2020-04-07 22:41:00.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/68/718268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/68/718268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/68/718268/1"}}},"commit":{"parents":[{"commit":"cd46003d839f520e0310494750c68c9d85596797","subject":"Merge \"Release python-zunclient 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd46003d839f520e0310494750c68c9d85596797"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:40:58.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:40:58.000000000","tz":-300},"subject":"Release final python-tackerclient for ussuri","message":"Release final python-tackerclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ibfc9469683fa9019d21c42b081681833193602cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cef33c45c66b89f5c3cb1cc3dd71d41b8b3dd8b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cef33c45c66b89f5c3cb1cc3dd71d41b8b3dd8b8"}]},"branch":"refs/heads/master"},"5a1fd28d9c9b13097c09a1ee694084367e111f2f":{"kind":"REWORK","_number":2,"created":"2020-04-09 15:31:06.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/68/718268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/68/718268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/68/718268/2"}}},"commit":{"parents":[{"commit":"cd46003d839f520e0310494750c68c9d85596797","subject":"Merge \"Release python-zunclient 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd46003d839f520e0310494750c68c9d85596797"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:40:58.000000000","tz":-300},"committer":{"name":"dharmendra kushwaha","email":"dharmendra.kushwaha@gmail.com","date":"2020-04-09 15:30:59.000000000","tz":0},"subject":"Release final python-tackerclient for ussuri","message":"Release final python-tackerclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ibfc9469683fa9019d21c42b081681833193602cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nCo-Authored-By: Dharmendra Kushwaha \u003cdharmendra.kushwaha@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a1fd28d9c9b13097c09a1ee694084367e111f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a1fd28d9c9b13097c09a1ee694084367e111f2f"}]},"branch":"refs/heads/master"},"dbae8ebecfee82aad2ddb38e60284ca1ef3b79f0":{"kind":"REWORK","_number":3,"created":"2020-04-10 01:42:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/68/718268/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/68/718268/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/68/718268/3"}}},"commit":{"parents":[{"commit":"cd46003d839f520e0310494750c68c9d85596797","subject":"Merge \"Release python-zunclient 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd46003d839f520e0310494750c68c9d85596797"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:40:58.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 01:42:29.000000000","tz":-300},"subject":"Release final python-tackerclient for ussuri","message":"Release final python-tackerclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ibfc9469683fa9019d21c42b081681833193602cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nCo-Authored-By: Dharmendra Kushwaha \u003cdharmendra.kushwaha@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dbae8ebecfee82aad2ddb38e60284ca1ef3b79f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dbae8ebecfee82aad2ddb38e60284ca1ef3b79f0"}]},"branch":"refs/heads/master"},"27eeef91a14567966c7f8da9dbb35e15e3d3a9c2":{"kind":"REWORK","_number":4,"created":"2020-04-10 07:49:24.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/68/718268/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/68/718268/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/68/718268/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/68/718268/4"}}},"commit":{"parents":[{"commit":"cd46003d839f520e0310494750c68c9d85596797","subject":"Merge \"Release python-zunclient 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd46003d839f520e0310494750c68c9d85596797"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-07 22:40:58.000000000","tz":-300},"committer":{"name":"dharmendra kushwaha","email":"dharmendra.kushwaha@gmail.com","date":"2020-04-10 07:49:08.000000000","tz":0},"subject":"Release final python-tackerclient for ussuri","message":"Release final python-tackerclient for ussuri\n\nFinal ussuri client lib release for the April 9 deadline.\n\nPlease leave +1 if we are all set and can proceed. Leave -1 if we\nneed to wait a little bit longer for something final to land,\nthen update it with a better commit hash to use for the release\nwhen ready.\n\nPatches with no response will be approved a few days after the\ndeadline.\n\nChange-Id: Ibfc9469683fa9019d21c42b081681833193602cf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nCo-Authored-By: Dharmendra Kushwaha \u003cdharmendra.kushwaha@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/27eeef91a14567966c7f8da9dbb35e15e3d3a9c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/27eeef91a14567966c7f8da9dbb35e15e3d3a9c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
