)]}'
{"id":"openstack%2Freleases~718456","triplet_id":"openstack%2Freleases~master~Iee562d5b3330da9ea2860f6b46d04c64829871c7","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Iee562d5b3330da9ea2860f6b46d04c64829871c7","subject":"Cycle highlights for ironic-inspector","status":"MERGED","created":"2020-04-08 13:52:53.000000000","updated":"2020-04-10 17:16:42.000000000","submitted":"2020-04-10 00:22:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718456-1586478171717-a17bf472","meta_rev_id":"dbf5ff47fbcfa96560a48ef5dab886201eda6fd0","_number":718456,"virtual_id_number":718456,"owner":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-09 20:30:33.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 00:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-10 00:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-09 07:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-10 00:11:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-09 07:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 07:39:38.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-09 20:30:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 00:11:20.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-04-10 00:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"574e329c086712f82c02a791ce4e96b0c826c79f","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2020-04-08 13:52:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10282c1c120c8869d93ae839899c5852a736df73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:09:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17de875e57b6485992a3693c4d11bab8 : SUCCESS in 10m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9fe3245a8074fccad8f75fcb23fbe0c : SUCCESS in 6m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/102c6945e9f74a4bbddf1049f1a6c9bf : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1f17626f1dd602fbcbc4dafe63e0058ec4381821","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-09 07:39:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"70dd931521a92f84ea67b087f8f1fb42831df847","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 07:43:20.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"f3945ea37cf030d2fe91521553cf73e628cd5415","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 20:30:33.000000000","message":"Patch Set 1:\n\nCycle highlights are by team, not by deliverable. Not sure this is wanted. Not really clear without the context of seeing it here what this is actually referring to and what that would mean to end users and consumers of those highlight notes.","accounts_in_message":[],"_revision_number":1},{"id":"e0b0883d6fc0ff4683eb7ada6132ef92c4fc7801","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-04-10 00:11:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79e7988e695267d13be5537e50334fab0e844979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:11:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1b832fed8ecfe8dcbcb2f35e20072ee5b4e551e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:22:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4cffaf344cc4c2dae22fa972d47a601 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04361c1bf05407d9733c758bf65ead6 : SUCCESS in 9m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58abf30b53c44fc9843ff610e45d34ae : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4c6b812b17d63e67b652ee08a4f6519681591e1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 00:22:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a04fffe8fa30087bd3fd730703bc93737737f47","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-10 16:54:16.000000000","message":"Patch Set 1:\n\n\u003e Cycle highlights are by team, not by deliverable.\n\nOh. Then this is one of the things we want to improve. *makes a mental note to bring it on the thread about ironic future*","accounts_in_message":[],"_revision_number":1},{"id":"6db2e9a190f519b347afb6811cf28fe2faaaf3ac","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 16:57:25.000000000","message":"Patch Set 1:\n\nIf you haven\u0027t seen it, this is the generated output that the marketing folks use:\n\nhttps://releases.openstack.org/ussuri/highlights.html\n\nThis is just to help them as raw input into what they want to ultimately message for the coordinated release. This isn\u0027t like release notes (that\u0027s already well covered), so getting into the detail of what each deliverable includes is too granular for the target audience and purpose.","accounts_in_message":[],"_revision_number":1},{"id":"fbbac3402ae0cdfe2dfc180e3ea809d1f4f9eb3c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-10 16:59:50.000000000","message":"Patch Set 1:\n\nMaybe then separate the highlights from a deliverable? It feels weird to put ironic-inspector highlights into the ironic developer given that the two services are quite independent.","accounts_in_message":[],"_revision_number":1},{"id":"dbf5ff47fbcfa96560a48ef5dab886201eda6fd0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 17:16:42.000000000","message":"Patch Set 1:\n\n\u003e Maybe then separate the highlights from a deliverable? It feels\n \u003e weird to put ironic-inspector highlights into the ironic developer\n \u003e given that the two services are quite independent.\n\nYeah, it is a bit odd. The foundation people asked for help getting a way to collect this from the teams. This was the best mechanism we had without introducing an entirely new set of processes for everyone to follow.\n\nPrior to this, direct calls were scheduled with each PTL to manually collect this type of information, which was obviously very time consuming. So although it isn\u0027t perfect, it has gone a long way to make it easier for everyone involved to get at least some of the right information into marketing hands to make sure what they are sharing publicly accurately reflects what the community has done.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9dbb848f36c21eb0e5e617eaa5dff3aaea57866","revisions":{"e9dbb848f36c21eb0e5e617eaa5dff3aaea57866":{"kind":"REWORK","_number":1,"created":"2020-04-08 13:52:53.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/56/718456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/718456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/718456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/718456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/718456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/718456/1"}}},"commit":{"parents":[{"commit":"346129e7994e5282d9ed73c427cd079d1639b986","subject":"Merge \"Bump vitrage version to 7.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/346129e7994e5282d9ed73c427cd079d1639b986"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-04-08 13:51:25.000000000","tz":120},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2020-04-08 13:51:25.000000000","tz":120},"subject":"Cycle highlights for ironic-inspector","message":"Cycle highlights for ironic-inspector\n\nChange-Id: Iee562d5b3330da9ea2860f6b46d04c64829871c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9dbb848f36c21eb0e5e617eaa5dff3aaea57866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9dbb848f36c21eb0e5e617eaa5dff3aaea57866"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
