)]}'
{"id":"openstack%2Freleases~718623","triplet_id":"openstack%2Freleases~master~I851a10e7fcbbd1f33c0e970b7a7a3806369ed80a","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I851a10e7fcbbd1f33c0e970b7a7a3806369ed80a","subject":"Release ironic-ui 3.5.5 for Train","status":"MERGED","created":"2020-04-09 09:17:37.000000000","updated":"2020-04-09 20:43:25.000000000","submitted":"2020-04-09 20:43:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718623-1586465005206-d5dcdf52","meta_rev_id":"2fdeba951600cd3b133861bea50e0e4ffd7a3550","_number":718623,"virtual_id_number":718623,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 20:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-09 20:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-09 10:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-04-09 10:21:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-09 20:28:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-09 10:22:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 10:27:23.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-09 20:28:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-09 20:43:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15d8fa9e6fa2e1a19eb5459ab4dcdaa5f01f32cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 09:17:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"542f1e0f086afab9ed95a82d9ce4ee9bbaf76d13","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:19:49.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"97259fd14170f6880445e5da71687107191813d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:32:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e803923b7b4bb19b6ede196b422d03 : SUCCESS in 12m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b1b3b43914e4e059520e14ef5653476 : SUCCESS in 10m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/80f8349d6c0940109e1d30b056ccb869 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"20b012e83e3d56f59b7a7005540ae509e5b1672d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-09 10:19:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"15607ccbbed931a669507e2671f90c36f8f0e070","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 10:21:22.000000000","message":"Patch Set 2: Code-Review+1\n\nOops, thanks","accounts_in_message":[],"_revision_number":2},{"id":"9678bc40cc2318b90bbdd567aeda072e70109729","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:22:29.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9ce3cd70933d55d77873a7686bfe543a9b6b0db0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-09 10:27:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"44e380225b48719fad06df91f63d53d7f47299b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:37:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/537a2cedb66e4633bc6fa0815a8ef88f : SUCCESS in 12m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86078e3dcd934cf39bb584ac8b7a8364 : SUCCESS in 9m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/865a1b1c66a14470a1e53704d9d4cc14 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e6a4052f6e9052852deb79bb43c2bbd7637c9671","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-09 20:28:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"890952a1762f6161c8c34621c36c041cdd8fc845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:29:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c22b8c91b93e43b41fcac923e80613df10213e62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:43:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d33bdabdb94b89a36e0c8771ceb447 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9773769c98824bccae7c71376b2936ad : SUCCESS in 12m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e5e7b32316e483a8ca24054604a0ffe : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2fdeba951600cd3b133861bea50e0e4ffd7a3550","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:43:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e169d343fd816cc50dee527526ed80ce47ebeed","revisions":{"192c04680c7f4aaf12cbe768dddb49aca369ec7d":{"kind":"REWORK","_number":1,"created":"2020-04-09 09:17:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/718623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/718623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/718623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/718623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/718623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/718623/1"}}},"commit":{"parents":[{"commit":"3dca58436daa60ff1de8db23cfbabbf0e837a41f","subject":"Merge \"Release final python-brick-cinderclient-ext for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3dca58436daa60ff1de8db23cfbabbf0e837a41f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-09 09:17:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-09 09:17:27.000000000","tz":60},"subject":"Release ironic-ui 3.5.4 for Train","message":"Release ironic-ui 3.5.4 for Train\n\nChange-Id: I851a10e7fcbbd1f33c0e970b7a7a3806369ed80a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/192c04680c7f4aaf12cbe768dddb49aca369ec7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/192c04680c7f4aaf12cbe768dddb49aca369ec7d"}]},"branch":"refs/heads/master"},"1e169d343fd816cc50dee527526ed80ce47ebeed":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-09 10:19:49.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/23/718623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/718623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/718623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/718623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/718623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/718623/2"}}},"commit":{"parents":[{"commit":"3dca58436daa60ff1de8db23cfbabbf0e837a41f","subject":"Merge \"Release final python-brick-cinderclient-ext for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3dca58436daa60ff1de8db23cfbabbf0e837a41f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-09 09:17:27.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-09 10:19:31.000000000","tz":0},"subject":"Release ironic-ui 3.5.5 for Train","message":"Release ironic-ui 3.5.5 for Train\n\nChange-Id: I851a10e7fcbbd1f33c0e970b7a7a3806369ed80a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e169d343fd816cc50dee527526ed80ce47ebeed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e169d343fd816cc50dee527526ed80ce47ebeed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
