)]}'
{"id":"openstack%2Freleases~718855","triplet_id":"openstack%2Freleases~master~Iaf565885aae6b01ea671c60855c5acbfcdd8c5e1","project":"openstack/releases","branch":"master","topic":"qinling","hashtags":[],"change_id":"Iaf565885aae6b01ea671c60855c5acbfcdd8c5e1","subject":"Stable branches for qinling libs","status":"MERGED","created":"2020-04-10 01:39:13.000000000","updated":"2020-04-11 18:40:59.000000000","submitted":"2020-04-11 18:40:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718855-1586630459843-1d1e633c","meta_rev_id":"f16a1effbcdd95d22cc1fe65e6b87bcdccb5fe15","_number":718855,"virtual_id_number":718855,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-11 18:40:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-11 18:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-11 18:25:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-11 18:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6df4884a86aa1ea3ef3b1d1953f1aea0e3c310df","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 01:39:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3b7335405025ae8a39742563b9375686d0e6f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 01:50:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da5cd4655ae45d9a717b4adb6d1f9ce : SUCCESS in 10m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e272785ef5fd41789997cf8148b03d9b : SUCCESS in 6m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/003a10b2dd8f4a04ad262455068c0022 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"83ebff041596872abb5b217df49596fcb926a8a9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-11 18:25:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a57ccd5685993e651472fc40e98edfe24785140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 18:25:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02828fd55252584e3c19e4b4b6f43279b823ffb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 18:40:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d1162fa3fce8a5f02d89c4d39a3135bcd23196f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 18:40:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1436eae61e424f3f9f46c46666383e15 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e10217db460400ea9476b7ceed0b935 : SUCCESS in 9m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2b3c9c6b6414b8d97db565fe2e3bfb3 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f96b81557025e683fdf3d8e8fb5c4b082b2b0200","revisions":{"f96b81557025e683fdf3d8e8fb5c4b082b2b0200":{"kind":"REWORK","_number":1,"created":"2020-04-10 01:39:13.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/55/718855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/718855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/718855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/718855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/718855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/718855/1"}}},"commit":{"parents":[{"commit":"8ad2dacba30210df01b9e41156348feaddc15332","subject":"Merge \"Release final python-cyborgclient for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ad2dacba30210df01b9e41156348feaddc15332"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 01:39:09.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-10 01:39:09.000000000","tz":-300},"subject":"Stable branches for qinling libs","message":"Stable branches for qinling libs\n\nChange-Id: Iaf565885aae6b01ea671c60855c5acbfcdd8c5e1\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f96b81557025e683fdf3d8e8fb5c4b082b2b0200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f96b81557025e683fdf3d8e8fb5c4b082b2b0200"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
