)]}'
{"id":"openstack%2Freleases~719097","triplet_id":"openstack%2Freleases~master~Ieb4eb7e695c932262965e812e39ce211f5b8c67e","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ieb4eb7e695c932262965e812e39ce211f5b8c67e","subject":"Octavia: EOL Ocata and Pike","status":"MERGED","created":"2020-04-10 20:12:33.000000000","updated":"2020-06-11 21:07:02.000000000","submitted":"2020-06-11 21:07:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719097-1591909622056-4874c0de","meta_rev_id":"3f9ef48057c0ab35b454576b97ee79ab34a405af","_number":719097,"virtual_id_number":719097,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 21:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-11 20:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-03 18:21:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-04-24 07:49:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-04-24 10:25:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-04-17 17:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-11 20:57:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-04-23 18:03:55.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-24 07:52:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 17:19:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-23 18:03:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-04-24 07:49:58.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-24 10:25:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-06-11 20:57:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-11 21:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc97cf072d5e39471be647d605f070c08cfda5b0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-10 20:12:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"643804afed51c24b53db31c492492d6c1a2a0d0d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 20:13:42.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"3240098ee56f4b8715db536b5042bbb44892e298","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 20:21:50.000000000","message":"Patch Set 1:\n\nThanks Adam. Is there a link to the announcement on openstack-discuss planning on doing this? We just need to make sure it\u0027s been 6 months [0].\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance","accounts_in_message":[],"_revision_number":1},{"id":"38241def04f92fc8e4e1b19237d78d6a89baa8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 20:26:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89945f7fcd794daf96d901600a9a37e3 : SUCCESS in 9m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5deeb4f56844de68e6937a7da5dcf28 : SUCCESS in 4m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bae039504ae1421c8f293077d7b5ea8c : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b6af0a9ff84e96faf351b7b79ee7815313e8e3eb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-14 07:17:03.000000000","message":"Patch Set 1: Code-Review+1\n\nChanges LGTM, \n\n@Sean: I suppose you asking for this mail [1]?\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014057.html","accounts_in_message":[],"_revision_number":1},{"id":"1f17bf4590ad20439fd7603a0d0687c89e9dde43","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-14 09:48:22.000000000","message":"Patch Set 1:\n\nThanks Adam, one question below, otherwise the patch looks OK to me: Pike and Ocata branches of Octavia had the last patch merged exactly 6 months ago, nothing after that.\n\nOne more question though, according to the EOL process [1]:\n\n\"Remove any related zuul jobs that are defined in other repositories and not needed anymore\"\n\nSo is there any zuul job defined outside of openstack/octavia or openstack/python-octaviaclient that is used in octavia\u0027s zuul jobs, which can be removed (i.e. not used in other places/repositories/branches) after Pike and Ocata move to EOL of these branches?\n\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#end-of-life","accounts_in_message":[],"_revision_number":1},{"id":"c63fd1946c7537b1f9511ef50a01a4d6b6aeef6f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-17 17:19:08.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM so +2","accounts_in_message":[],"_revision_number":1},{"id":"cba599ce1a18a3eb0966e9f6ab8b5b0c2b7ad54b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 17:26:48.000000000","message":"Patch Set 1: -PTL-Approved","accounts_in_message":[],"_revision_number":1},{"id":"3b6184b9be6607838ea14b47e6e4795b140f8d35","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-24 07:49:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6bb44b45d4b67aa4e1fcf299e0ee93f2f9c53de9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 07:52:09.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"5e698cbdcaa9c6bb48c91b9ad2fae533250ae932","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-24 10:25:34.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e Thanks Adam, one question below, otherwise the patch looks OK to\n \u003e me: Pike and Ocata branches of Octavia had the last patch merged\n \u003e exactly 6 months ago, nothing after that.\n \u003e \n \u003e One more question though, according to the EOL process [1]:\n \u003e \n \u003e \"Remove any related zuul jobs that are defined in other\n \u003e repositories and not needed anymore\"\n \u003e \n \u003e So is there any zuul job defined outside of openstack/octavia or\n \u003e openstack/python-octaviaclient that is used in octavia\u0027s zuul jobs,\n \u003e which can be removed (i.e. not used in other places/repositories/branches)\n \u003e after Pike and Ocata move to EOL of these branches?\n \u003e \n \u003e [1] https://docs.openstack.org/project-team-guide/stable-branches.html#end-of-life\n\nTo answer my own question:\n\nI went through quickly the zuul jobs and it seems there are two kinds of jobs:\n* pretty general jobs, that are used by most of the openstack projects even in master\n* octavia or neutron-lbaas specific jobs that are defined within these repositories and used by both projects.\n\nI haven\u0027t found any other job that is outside of Octavia or Neutron-lbaas repositories and not used in other places. So it seems OK to me: no leftover jobs found elsewhere.","accounts_in_message":[],"_revision_number":1},{"id":"24855d9df84dac6cd918f22d7aacb5dd053c5e22","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-03 18:21:48.000000000","message":"Patch Set 1: Code-Review+1\n\nSo, we should be good to go here, yes? We would like to have this merged whenever possible! :)","accounts_in_message":[],"_revision_number":1},{"id":"06a514d434dc6c6640a445f12d921c404fd26dc4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-11 20:51:30.000000000","message":"Patch Set 1:\n\nBump!","accounts_in_message":[],"_revision_number":1},{"id":"91d64946d8f82b37585916df333efc615db5ca2e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-11 20:57:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nShould be good to go.","accounts_in_message":[],"_revision_number":1},{"id":"4600bcb44d9550881542623c34a3344fc4fd8ade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 20:57:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a27e3b90f1481dd3af117cf8032aae598fb5d32e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 21:07:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818c57b055dd44ddac6a1a35b631e241 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3921d4d681a439795bb0435706328f0 : SUCCESS in 8m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ee89fa16ed7422d809c4d721aa43ba5 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3f9ef48057c0ab35b454576b97ee79ab34a405af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 21:07:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1038fcbdd7a6d22b13623c609a448244a92ee253","revisions":{"1038fcbdd7a6d22b13623c609a448244a92ee253":{"kind":"REWORK","_number":1,"created":"2020-04-10 20:12:33.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/97/719097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/719097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/719097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/719097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/719097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/719097/1"}}},"commit":{"parents":[{"commit":"aaeaf0c04aef232dd47f9b944314e5b3388b82bc","subject":"Merge \"[manila] Extend cycle highlights for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aaeaf0c04aef232dd47f9b944314e5b3388b82bc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-10 20:11:25.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-10 20:11:25.000000000","tz":-420},"subject":"Octavia: EOL Ocata and Pike","message":"Octavia: EOL Ocata and Pike\n\nThis should have been done a bit ago.\n\nChange-Id: Ieb4eb7e695c932262965e812e39ce211f5b8c67e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1038fcbdd7a6d22b13623c609a448244a92ee253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1038fcbdd7a6d22b13623c609a448244a92ee253"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
