)]}'
{"id":"openstack%2Freleases~720854","triplet_id":"openstack%2Freleases~master~I6c6eb2297159bd9cc6675d3ec86c71ed2f7b79e9","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I6c6eb2297159bd9cc6675d3ec86c71ed2f7b79e9","subject":"Swift 2.25.0 release and stable/ussuri branch","status":"MERGED","created":"2020-04-18 01:31:50.000000000","updated":"2020-04-20 09:50:47.000000000","submitted":"2020-04-20 09:50:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720854-1587376247991-a68383b8","meta_rev_id":"567b2c341968511178c725a91a1294cfe055869a","_number":720854,"virtual_id_number":720854,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 09:50:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-19 05:17:28.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-04-20 09:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-04-20 09:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-04-20 09:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 09:13:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-20 09:22:33.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-20 09:50:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a38d33c5d757f1e9cbab34b16d3f40e62ad72bb2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-18 01:31:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef0bab706fb0fcb8f1411b3b832651e75bccd541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 01:39:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3510f513267d47b0b47197b1ebdb757a : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7dcbdaa27fc14da0a8a9978df60e6417 : FAILURE in 6m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3e3e084b8c7442c08171ef2090588d1f : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1abc7bb3383ea6275f5275a94091c9b7033e9c74","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-19 05:17:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"552ea5966a707bfacf72bda7d6fd018ee1fffca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 05:25:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfd19d2a5dc45e4a085ca24d5e630fc : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9614901ea173491f91e9a56af57f844b : SUCCESS in 6m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fe6cfe8d0a8046a8b2f289952b499184 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fb4317c961128f55111efc9b4969973f903f1948","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-20 09:13:15.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, waiting for PTL approval","accounts_in_message":[],"_revision_number":1},{"id":"d2a0e98086766569ad205c9b6d7acfb09a0cde40","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-20 09:22:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTim is PTL/Liaison, just not on that email","accounts_in_message":[],"_revision_number":1},{"id":"5160dc84150faa733e260b3c8de20fbd226f72cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 09:23:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03bd69d0218827cc3dc7f4552860d8615e39852e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 09:50:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"567b2c341968511178c725a91a1294cfe055869a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 09:50:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c92cbcfbb72441ef87ac2d5721a51741 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e34e6ee56344652a144fc8478871e1d : SUCCESS in 8m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb1fa4a448d5456caa0417f70eac22a7 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f18ea3bed311c784d3d6034ae67b5a4ce212432","revisions":{"0f18ea3bed311c784d3d6034ae67b5a4ce212432":{"kind":"REWORK","_number":1,"created":"2020-04-18 01:31:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/720854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/720854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/720854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/720854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/720854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/720854/1"}}},"commit":{"parents":[{"commit":"f8d756bf5bc79a96b9f4460517046ae5c5fdf419","subject":"Merge \"Create new release 15.5.0 for puppet-magnum in train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f8d756bf5bc79a96b9f4460517046ae5c5fdf419"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-17 23:59:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-18 01:31:29.000000000","tz":-420},"subject":"Swift 2.25.0 release and stable/ussuri branch","message":"Swift 2.25.0 release and stable/ussuri branch\n\nDepends-On: https://review.opendev.org/#/c/720851/\nChange-Id: I6c6eb2297159bd9cc6675d3ec86c71ed2f7b79e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f18ea3bed311c784d3d6034ae67b5a4ce212432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f18ea3bed311c784d3d6034ae67b5a4ce212432"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
