)]}'
{"id":"openstack%2Freleases~721095","triplet_id":"openstack%2Freleases~master~I8f33d4c0bf5fcba203d410cd021c548219a757ec","project":"openstack/releases","branch":"master","topic":"docs-cleanup","hashtags":[],"change_id":"I8f33d4c0bf5fcba203d410cd021c548219a757ec","subject":"Update docs building, Python 3 cleanup","status":"MERGED","created":"2020-04-19 15:54:03.000000000","updated":"2020-04-20 13:08:34.000000000","submitted":"2020-04-20 13:08:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721095-1587388114654-77dd523e","meta_rev_id":"810ddd0b9d9b41834963e6d73ce218749f4ac584","_number":721095,"virtual_id_number":721095,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-20 12:18:05.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 13:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-20 09:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-04-20 09:17:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-20 09:27:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 09:17:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-20 09:27:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-20 13:08:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af0ae8d18ac1cb0698990c4ca1b146aee00fb0a3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-19 15:54:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af6d76f57b880c1452d66b3ee45aeb6016335771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 16:31:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab1cda24da549ecb84cefe67d81abd6 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80c257bc1c834d99bcca960c883b2bcb : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3962a72814a487d9c01d851e5bf5930 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b92083b77cd419b95cde6fc6870185f : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7d40af252a4ed8acfd78c0f62813cc : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/652b43c7be1d443a9dcfe13d0fe752d2 : SUCCESS in 4m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a096fd5c69d949a8ac3fb774083a74ea : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1b5c65285630640cc1179dc67fa920d08b93b31e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-20 09:15:21.000000000","message":"Patch Set 1: Code-Review-1\n\nYou missed:\n\nhttps://github.com/openstack/releases/blob/master/test-requirements.txt#L17\n\nIt should be dropped now.\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"b2108d81ea4239ba0da8aeab6a795ae5615f1210","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 09:16:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4028aac83a63eb67df9ea6af4b2807a6b366876f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 09:17:20.000000000","message":"Patch Set 2:\n\n\u003e You missed:\n \u003e \n \u003e https://github.com/openstack/releases/blob/master/test-requirements.txt#L17\n \u003e \n \u003e It should be dropped now.\n\nIndeed, fixed now. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"a18966cd408673ac76b155da228d58f3a6627cd1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-20 09:17:59.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"172f4c32e07c07d2c136bc28a13e45dc49007747","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-20 09:27:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"89aba6a6583e4b8cd4eca34c6a277592ebfbebe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 10:02:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14b493e618674e68ac1f81d59ca1dcc9 : POST_FAILURE in 10m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1224085f3ea14cfd93a22126208bd6d3 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b4501d991d74816a1ecb89385565889 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f10a29546684e5bbb9fff7119f93af8 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3b4c10045a4c1b8f23c5b20af8fd95 : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b86427e95b04ac08c0991dacbee9b2f : SUCCESS in 7m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/54f8538cb42e49dc80140f590fa96bbe : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"80393431c771b99bee45f90e85fc58ccbc4658aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 10:07:08.000000000","message":"Patch Set 2:\n\nrecheck - timeout","accounts_in_message":[],"_revision_number":2},{"id":"543d5112f802706a5067bed413f582ce7c9ef6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 12:08:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2b50ad994b44238f0aba59536aac60 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ce542f832614dd7a008cdd79ab4d7e5 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b683aaa978e34914909ecbdbb7b314fa : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5336527d8d3544d58899ab94a58585bd : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dba3422a0784f3aa93fa78fb9c20d68 : SUCCESS in 7m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35fd50da2de445079bb8f665c28a50f9 : POST_FAILURE in 9m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/05384bf1b64c4547adff2269aad62c9e : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2080ff20774a9c2fb9838d4f32776a3c02a7290c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 12:13:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"afd91d4dbdc99c52ffca739d8e39e630c2081b85","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 12:18:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5aa3a5f5b543f4761e681f88078dd7dc4be18283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 12:56:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a68fcb423d741778db36fc050aac2b6 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1b0b088d65c440ba3bbf2e5d849f1b7 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/043ec3dbe41a44dc9aa7a18723b782aa : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f29ffe88af3e4f29946c01202732663f : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d614c3cf11514015b5c33cb87841367e : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f22eb3be4a9349ebb7b8b0bb4b95a679 : SUCCESS in 6m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a4bc6594a4b745fa8b52277ec8c91d48 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b52233e6f2e7e682631db1bd17d560ceb87130dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 12:56:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72f20136368723e0cb84fa582c4b3e8c16a66931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 13:08:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b6089365034a03afa4690604438c23 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aedbe53f0624c2eaf9f5465c3c41dd9 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0eb63cf163b74d7b9ff2e8923bf3190c : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9066788f48040589af49cc92fdf094f : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dde6390e5daf4362a9bc7d0ca07cd4f7 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ea55ad4106d93d6d69cfccf1fb34714227db6151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 13:08:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c314146776e907ded7dbc33a976d1fed519c7ba","revisions":{"c61d1f3826f3f1a002a46bdcbce7da3310c2ddc8":{"kind":"REWORK","_number":1,"created":"2020-04-19 15:54:03.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/721095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/721095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/721095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/721095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/721095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/721095/1"}}},"commit":{"parents":[{"commit":"29d7290c61026469029428557bcaed992667ecbe","subject":"Merge \"Release final python-monascaclient for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/29d7290c61026469029428557bcaed992667ecbe"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 15:47:22.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 15:53:56.000000000","tz":120},"subject":"Update docs building, Python 3 cleanup","message":"Update docs building, Python 3 cleanup\n\nUpdate docs building:\n* Remove unneeded doc sections from setup.cfg\n* Change constraints to use published document, use\n  new variable\n* Import mock from unittest, remove imports from future\n\nChange-Id: I8f33d4c0bf5fcba203d410cd021c548219a757ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c61d1f3826f3f1a002a46bdcbce7da3310c2ddc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c61d1f3826f3f1a002a46bdcbce7da3310c2ddc8"}]},"branch":"refs/heads/master"},"0c314146776e907ded7dbc33a976d1fed519c7ba":{"kind":"REWORK","_number":2,"created":"2020-04-20 09:16:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/721095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/721095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/721095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/721095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/721095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/721095/2"}}},"commit":{"parents":[{"commit":"29d7290c61026469029428557bcaed992667ecbe","subject":"Merge \"Release final python-monascaclient for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/29d7290c61026469029428557bcaed992667ecbe"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-19 15:47:22.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-20 09:16:46.000000000","tz":120},"subject":"Update docs building, Python 3 cleanup","message":"Update docs building, Python 3 cleanup\n\nUpdate docs building:\n* Remove unneeded doc sections from setup.cfg\n* Change constraints to use published document, use\n  new variable\n* Import mock from unittest, remove imports from future\n\nChange-Id: I8f33d4c0bf5fcba203d410cd021c548219a757ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c314146776e907ded7dbc33a976d1fed519c7ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c314146776e907ded7dbc33a976d1fed519c7ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
