)]}'
{"id":"openstack%2Freleases~721381","triplet_id":"openstack%2Freleases~master~I87bd5fbc41d51a14f738991bd446ef8e0a507549","project":"openstack/releases","branch":"master","topic":"c-w-i-ussuri","hashtags":[],"change_id":"I87bd5fbc41d51a14f738991bd446ef8e0a507549","subject":"Release cloudkitty for ussuri","status":"MERGED","created":"2020-04-21 01:41:42.000000000","updated":"2020-04-24 18:05:56.000000000","submitted":"2020-04-24 18:05:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721381-1587751556770-792c2e80","meta_rev_id":"3181465a9ab4a48049b6248a9bab222e3dad2cda","_number":721381,"virtual_id_number":721381,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 18:05:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-24 17:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-24 17:48:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-04-23 19:12:14.000000000","_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 19:12:14.000000000","updated_by":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"reviewer":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"state":"REVIEWER"},{"updated":"2020-04-24 18:05:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d7679e7c0346f9e46c62bd5b5fd81a6fdc39256","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-21 01:41:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86501f2d9e605b3f1a79a1e4982ffb596067e7b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 01:50:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653fa2acc8e645e1848d28fc3ca76245 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23606eb58bee4e2ab9804b9c1fb4b7a8 : SUCCESS in 4m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/733c0c3503b247929dc90d72ac72f963 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"28b8cd42a15998a0f86919dca044d74448813bc1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 17:48:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPast deadline. No new patches have merged, and doesn\u0027t look like any are about to.","accounts_in_message":[],"_revision_number":1},{"id":"5695d225e43ffb8c06554c783e9246da338e0649","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 17:49:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9367def3fb46411e9da4ed03a0f993cd2186f624","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:05:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88d285bdfc3443f2a829efce41a3fd90 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc57d7e2244e468baa3623f1d42c7355 : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6784388fe31040eb8c19424cf830bac6 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"934fd41ed55c30e167b8a2779f38e715e5bfb60b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:05:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"407c1529f7474bd0a386753b716f961100817a36","revisions":{"407c1529f7474bd0a386753b716f961100817a36":{"kind":"REWORK","_number":1,"created":"2020-04-21 01:41:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/81/721381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/721381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/721381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/721381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/721381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/721381/1"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:41:39.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:41:39.000000000","tz":-300},"subject":"Release cloudkitty for ussuri","message":"Release cloudkitty for ussuri\n\nThis is the final release for cloudkitty for the ussuri\ncycle.\n\nPlease leave +1 if this is ready to go, or -1 if the team needs a\nlittle more time. If so, update this path with the new commit\nhash to use once ready.\n\nPatches with no response will be processed soon after the April\n23 deadline.\n\nChange-Id: I87bd5fbc41d51a14f738991bd446ef8e0a507549\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/407c1529f7474bd0a386753b716f961100817a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/407c1529f7474bd0a386753b716f961100817a36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
