)]}'
{"id":"openstack%2Freleases~721408","triplet_id":"openstack%2Freleases~master~I12639f34545db9b4f6f665c23b310aeb90c2c2b7","project":"openstack/releases","branch":"master","topic":"ussuri-rc","hashtags":[],"change_id":"I12639f34545db9b4f6f665c23b310aeb90c2c2b7","subject":"Release heat for Ussuri","status":"MERGED","created":"2020-04-21 01:57:29.000000000","updated":"2020-04-27 13:48:05.000000000","submitted":"2020-04-27 13:48:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721408-1587995285046-f65c5a5b","meta_rev_id":"5fcd5b077befa4104e34f1ac519946cb0c5dadac","_number":721408,"virtual_id_number":721408,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 13:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-27 13:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-27 12:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-27 13:08:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-27 12:28:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 12:42:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-27 12:52:54.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-04-27 13:48:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cffff3030b8192ed278d77444607dd3d9e8d61e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-21 01:57:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9ae17cad770cc60749d7a0eb5d1d9faa3734a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:11:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a9abe9508b44f19df842ccc0ded9be : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7a93b31292c412494caef266e5680c8 : SUCCESS in 9m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7bfe78e0324f4796b223e9516eb1445f : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4a0ed9222bb19e97ec5d342456c15e41ae99dc5b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-23 12:42:22.000000000","message":"Patch Set 1: Code-Review+1\n\nPy2 support dropping LGTM to justify a major release.\nWaiting for PTL review.","accounts_in_message":[],"_revision_number":1},{"id":"bed1171a2c095657c9ff2c4c78502fb9d8ea4270","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-23 12:44:40.000000000","message":"Patch Set 1:\n\nMaybe it could be worth to wait/embed these changes [1] too (on ussuri), to completely drop the py27 support under this cycle.\n\n[1] https://review.opendev.org/#/q/project:openstack/heat+branch:master+topic:drop-six-and-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"5cd09a09631531416a81726dec09ccf6ad2aab93","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-23 21:24:47.000000000","message":"Patch Set 1:\n\n\u003e Maybe it could be worth to wait/embed these changes [1] too (on\n \u003e ussuri), to completely drop the py27 support under this cycle.\n \u003e \n \u003e [1] https://review.opendev.org/#/q/project:openstack/heat+branch:master+topic:drop-six-and-py27-support\n\nLooks like this already completed the goal. Additional six and other compatibility cleanup work doesn\u0027t need to be part of ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"16820ca5e093786eb8ef21465abd71863fe7d3cd","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-04-24 16:15:26.000000000","message":"Patch Set 1:\n\nwe do have at least two feature patches [1] and [2] on the way, so if we can wait for them till the last minute will be nice.\n[1] https://review.opendev.org/#/c/709129/4\n[2] https://review.opendev.org/#/c/680312/","accounts_in_message":[],"_revision_number":1},{"id":"29feaca88ac6f7e021ee988936219210b10c2428","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 16:40:28.000000000","message":"Patch Set 1:\n\n\u003e we do have at least two feature patches [1] and [2] on the way, so\n \u003e if we can wait for them till the last minute will be nice.\n \u003e [1] https://review.opendev.org/#/c/709129/4\n \u003e [2] https://review.opendev.org/#/c/680312/\n\nWell, that last minute was yesterday, but we can wait a little bit longer. :)","accounts_in_message":[],"_revision_number":1},{"id":"4aff4ad090a81ffa994dcdbf57473c939e7670f3","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-04-24 17:18:14.000000000","message":"Patch Set 1: Code-Review+1\n\nOkay, let\u0027s move those FFE to V cycle and call it Ussuri for heat here:)","accounts_in_message":[],"_revision_number":1},{"id":"c41af783f578d756b1e4b90899cf38be7f58849e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 17:28:23.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"dc8884c3553259893432635ed0fe17d871895a66","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-04-27 12:26:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"20d0a4c769115a6e663c7fe1d9c607805056cdf2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 12:28:05.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ff3af53d259e9d90a346b7da46caf9818f55aa24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 12:41:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10b68108f0841d1a5798dd24a8b01d7 : SUCCESS in 8m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5290a507fa85419080242a1e503085c5 : SUCCESS in 10m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d77ef977ca4a4d278d60b3340e1ac12c : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"eda1510a7a09e530aeeaff8c2a8494658df74e79","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-04-27 12:52:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"551ece82b830cc26b3c1ddf66559cec8961795bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 13:08:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2f6608265c92c94e11216703407d00dd9b035fce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:28:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a718884911bcafa9309b8a1b282e32c09ad266c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:48:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca72f6e176154987b8aa23624df0cc21 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2a2d120d444353aa8bc555e62388d0 : SUCCESS in 9m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a89f31c7b39745e4a46845074f64851c : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":2},{"id":"ef67d54e3da9ed36f908b90a72827d05b4187163","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:48:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cb5f645b1e1c448d0c6175fc7f69c56b1ce7948c","revisions":{"be2357d317ed540be55d092fea148599faceeda2":{"kind":"REWORK","_number":1,"created":"2020-04-21 01:57:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/08/721408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/721408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/721408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/721408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/721408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/721408/1"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:57:27.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:57:27.000000000","tz":-300},"subject":"Release heat for Ussuri","message":"Release heat for Ussuri\n\nRC1 release for heat. Please +1 if things are\nready to go, or -1 if the team needs a little more\ntime. If so, please update this patch with the new\ncommit hash to use when ready.\n\nIf no response, this patch will be approved soon\nafter the April 23 deadline.\n\nChange-Id: I12639f34545db9b4f6f665c23b310aeb90c2c2b7\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be2357d317ed540be55d092fea148599faceeda2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be2357d317ed540be55d092fea148599faceeda2"}]},"branch":"refs/heads/master"},"cb5f645b1e1c448d0c6175fc7f69c56b1ce7948c":{"kind":"REWORK","_number":2,"created":"2020-04-27 12:26:49.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/08/721408/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/08/721408/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/721408/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/721408/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/08/721408/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/08/721408/2"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:57:27.000000000","tz":-300},"committer":{"name":"Rico Lin","email":"rico.lin.guanyu@gmail.com","date":"2020-04-27 12:26:46.000000000","tz":0},"subject":"Release heat for Ussuri","message":"Release heat for Ussuri\n\nRC1 release for heat. Please +1 if things are\nready to go, or -1 if the team needs a little more\ntime. If so, please update this patch with the new\ncommit hash to use when ready.\n\nIf no response, this patch will be approved soon\nafter the April 23 deadline.\n\nChange-Id: I12639f34545db9b4f6f665c23b310aeb90c2c2b7\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb5f645b1e1c448d0c6175fc7f69c56b1ce7948c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb5f645b1e1c448d0c6175fc7f69c56b1ce7948c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
