)]}'
{"id":"openstack%2Freleases~721409","triplet_id":"openstack%2Freleases~master~Icc783a34409de20ee4938ee6bece22738a63977d","project":"openstack/releases","branch":"master","topic":"ussuri-rc","hashtags":[],"change_id":"Icc783a34409de20ee4938ee6bece22738a63977d","subject":"Release keystone for Ussuri","status":"MERGED","created":"2020-04-21 01:58:07.000000000","updated":"2020-04-24 08:39:47.000000000","submitted":"2020-04-24 08:39:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721409-1587717587672-a0c18567","meta_rev_id":"f26197114bc289c06846a23f298e61b42db46b58","_number":721409,"virtual_id_number":721409,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 08:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2020-04-24 08:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2020-04-24 08:29:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-23 21:48:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 11:39:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-23 17:27:44.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-04-24 08:29:54.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-24 08:39:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb800ea002c4b129dfee400dbe5780a6e3b8f540","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-21 01:58:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"340d77720fa6cb78ca6c4f4df988d17e9479deb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:05:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c503f109f8964677a1a14f168d35091a : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b12d36f1a9c489b9448f8af430e070d : SUCCESS in 6m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5ec19b25ed994e22b591151e4b4cd60e : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b6d06e8839e5a92e504fd244e66db810ce555c5d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-23 11:39:03.000000000","message":"Patch Set 1: Code-Review+1\n\nPy2 support seems dropped (should be double checked) LGTM for a major release.\n\nWaiting for PTL review.","accounts_in_message":[],"_revision_number":1},{"id":"84afefac0940f69611cee94c3849fe838d0f5122","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-04-23 17:27:44.000000000","message":"Patch Set 1: Code-Review-1\n\nNeeds https://review.opendev.org/#/c/721688 first","accounts_in_message":[],"_revision_number":1},{"id":"5f3e9b4ed7a787f8bdc6720999afa6d54c97f6b2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-23 21:22:23.000000000","message":"Patch Set 1:\n\n\u003e Needs https://review.opendev.org/#/c/721688 first\n\nLooks like this has now merged. Can we get an update with the hash you want to use?","accounts_in_message":[],"_revision_number":1},{"id":"e51fa09bb2d06bdefcda5bf7deb8d022cddb1741","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-04-23 21:46:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa73fa293fd1cdf612cb2bf2b64ce383e55599b5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:48:58.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"90d8b074cf07f97e219d123037721bdcb0d49bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:02:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c5f8ce9e384a5184e67e01a2927a61 : SUCCESS in 10m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37ed1ceef66149249d7dd01596c58fc1 : SUCCESS in 10m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/737d3320b50644079698cf49620e1e82 : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":2},{"id":"b8b7c3c0f300bcbad0c941e2abff21cb2f304b77","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-24 08:29:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"254a26cb9a9ebbb77ceca99cc99fb7074972d48f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 08:30:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9aeb638029b055c2900b96762b3b40c8c9d36590","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 08:39:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d43ed2d3bb3f670538c903dab174f00b560438fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 08:39:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0587b23604d0497e83b9fb3fe8f8faf0 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc6b162022f4e9bb2af7082139d24d1 : SUCCESS in 7m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6379817af674626a6254a9009548951 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e9ae0de3b22718d3ac0934a9ff43e20fb5964e6f","revisions":{"6bcc0fd3c1d740d0439905dcd38546d5fad3665a":{"kind":"REWORK","_number":1,"created":"2020-04-21 01:58:07.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/09/721409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/721409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/721409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/721409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/721409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/721409/1"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:58:05.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:58:05.000000000","tz":-300},"subject":"Release keystone for Ussuri","message":"Release keystone for Ussuri\n\nRC1 release for keystone. Please +1 if things are\nready to go, or -1 if the team needs a little more\ntime. If so, please update this patch with the new\ncommit hash to use when ready.\n\nIf no response, this patch will be approved soon\nafter the April 23 deadline.\n\nChange-Id: Icc783a34409de20ee4938ee6bece22738a63977d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bcc0fd3c1d740d0439905dcd38546d5fad3665a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bcc0fd3c1d740d0439905dcd38546d5fad3665a"}]},"branch":"refs/heads/master"},"e9ae0de3b22718d3ac0934a9ff43e20fb5964e6f":{"kind":"REWORK","_number":2,"created":"2020-04-23 21:46:44.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/09/721409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/721409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/721409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/721409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/721409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/721409/2"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:58:05.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2020-04-23 21:46:29.000000000","tz":-240},"subject":"Release keystone for Ussuri","message":"Release keystone for Ussuri\n\nRC1 release for keystone. Please +1 if things are\nready to go, or -1 if the team needs a little more\ntime. If so, please update this patch with the new\ncommit hash to use when ready.\n\nIf no response, this patch will be approved soon\nafter the April 23 deadline.\n\nChange-Id: Icc783a34409de20ee4938ee6bece22738a63977d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9ae0de3b22718d3ac0934a9ff43e20fb5964e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9ae0de3b22718d3ac0934a9ff43e20fb5964e6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
