)]}'
{"id":"openstack%2Freleases~721410","triplet_id":"openstack%2Freleases~master~I57d3aca465c14c238c28d5d2edcbc22259262a6d","project":"openstack/releases","branch":"master","topic":"ussuri-rc","hashtags":[],"change_id":"I57d3aca465c14c238c28d5d2edcbc22259262a6d","subject":"Release kuryr-kubernetes for Ussuri","status":"MERGED","created":"2020-04-21 01:58:24.000000000","updated":"2020-04-23 15:23:04.000000000","submitted":"2020-04-23 15:23:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721410-1587655384025-653b66dd","meta_rev_id":"4084aa0e7e4d9528ef2468b79c219e780686c524","_number":721410,"virtual_id_number":721410,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 15:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2020-04-23 12:45:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-04-23 15:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-04-23 14:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-04-23 15:12:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-23 12:46:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 12:45:31.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-04-23 14:19:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-23 15:12:48.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-23 15:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e8a83916ffcb4a53952dd4aa77cd0d89e484db7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-21 01:58:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b11f45a4bccccda1d0c672d1af91ced8de67349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:10:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4b1ad78cc24661bcb62cac8bb3e2d6 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09bdd69177a24f50b9a4375b1cc380e4 : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9e65f79c85464e66996c5b5b022e81be : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e83c832c61e810a825b021de33b9bcecd3676bac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-23 12:37:14.000000000","message":"Patch Set 1: Code-Review+1\n\nI don\u0027t seen a specific commit related to the py2 support dropping which could justify a major release.\n\nHowever, I seen many changes related to the compat with py3 and the drop of six, so I guess we could consider that this version broke the backward compat, so a major release LGTM.\n\nThoughts?\n \nAnyway, we waiting for PTL review.","accounts_in_message":[],"_revision_number":1},{"id":"56b17e6ffa9380cfa2ffc0eff2d65383f9b05088","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-23 12:42:48.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e I don\u0027t seen a specific commit related to the py2 support dropping\n \u003e which could justify a major release.\n \u003e \n \u003e However, I seen many changes related to the compat with py3 and the\n \u003e drop of six, so I guess we could consider that this version broke\n \u003e the backward compat, so a major release LGTM.\n \u003e \n \u003e Thoughts?\n \u003e \n \u003e Anyway, we waiting for PTL review.\n\nWhoops, somehow I missed this. I already created I7863cb7176e8e478b2adb72db2b3a849f369cf27, which bases from last-minute doc update to fulfill the contributor docs goal. The 2.0 also makes sense, we dropped Python 2 in Id9983d2fd83cef89e3198b2760816cf4a851008b.\n\nI\u0027ll abandon my patch and update this one.","accounts_in_message":[],"_revision_number":1},{"id":"2ca0a104c09e5faff9e06441d4d5543ba5d45e27","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-23 12:44:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e3a2620465c18b312226694044c6739d8217c5e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-04-23 12:45:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a0f545b7a33757b57bcf0a32d53cf03e2ad3803f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 12:46:48.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"4c3da93959f6cf18763fbc9e3dae71d32171e457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 12:58:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1d48fa5b4f4d63bd5c769a3c99f9fe : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6a8e8c1a03544a29af3bd9cf1846d5e : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/da62cf44d9fa45429790c2968f24fc53 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2},{"id":"8093c905b8d77c84461018b2fa370ecfaea535c6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-23 14:19:57.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e \u003e I don\u0027t seen a specific commit related to the py2 support\n \u003e dropping\n \u003e \u003e which could justify a major release.\n \u003e \u003e\n \u003e \u003e However, I seen many changes related to the compat with py3 and\n \u003e the\n \u003e \u003e drop of six, so I guess we could consider that this version broke\n \u003e \u003e the backward compat, so a major release LGTM.\n \u003e \u003e\n \u003e \u003e Thoughts?\n \u003e \u003e\n \u003e \u003e Anyway, we waiting for PTL review.\n \u003e \n \u003e Whoops, somehow I missed this. I already created I7863cb7176e8e478b2adb72db2b3a849f369cf27,\n \u003e which bases from last-minute doc update to fulfill the contributor\n \u003e docs goal. The 2.0 also makes sense, we dropped Python 2 in\n \u003e Id9983d2fd83cef89e3198b2760816cf4a851008b.\n \u003e \n \u003e I\u0027ll abandon my patch and update this one.\n\nAck, thanks for the heads up","accounts_in_message":[],"_revision_number":2},{"id":"9a601feefa6729b9fbd1741d9dfa0f2ed9be03d0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-23 15:12:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8c65616ce97735e8026d8ef74a3d1de10966f071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:13:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3cfed90610c9c675d30add6a4ac129b31630dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:23:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3f659af0262414f8af88f0115328a3e : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b688beb3beac4e92a5dd9c2700e79f86 : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1766db581a6443dd9ef1a858e6059d7a : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":2},{"id":"911c390bbac18f2f35e0ad28d853c9e9dc56382d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:23:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc1ebf3926980adf91187c154c1646c386fe39a0","revisions":{"dcd2ce1396228a206b4884928326bfd628490637":{"kind":"REWORK","_number":1,"created":"2020-04-21 01:58:24.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/10/721410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/10/721410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/721410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/721410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/721410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/10/721410/1"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:58:18.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:58:18.000000000","tz":-300},"subject":"Release kuryr-kubernetes for Ussuri","message":"Release kuryr-kubernetes for Ussuri\n\nRC1 release for kuryr-kubernetes. Please +1 if things are\nready to go, or -1 if the team needs a little more\ntime. If so, please update this patch with the new\ncommit hash to use when ready.\n\nIf no response, this patch will be approved soon\nafter the April 23 deadline.\n\nChange-Id: I57d3aca465c14c238c28d5d2edcbc22259262a6d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dcd2ce1396228a206b4884928326bfd628490637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dcd2ce1396228a206b4884928326bfd628490637"}]},"branch":"refs/heads/master"},"cc1ebf3926980adf91187c154c1646c386fe39a0":{"kind":"REWORK","_number":2,"created":"2020-04-23 12:44:56.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/10/721410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/10/721410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/721410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/721410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/10/721410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/10/721410/2"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:58:18.000000000","tz":-300},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-04-23 12:44:44.000000000","tz":120},"subject":"Release kuryr-kubernetes for Ussuri","message":"Release kuryr-kubernetes for Ussuri\n\nRC1 release for kuryr-kubernetes. Please +1 if things are\nready to go, or -1 if the team needs a little more\ntime. If so, please update this patch with the new\ncommit hash to use when ready.\n\nIf no response, this patch will be approved soon\nafter the April 23 deadline.\n\nChange-Id: I57d3aca465c14c238c28d5d2edcbc22259262a6d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc1ebf3926980adf91187c154c1646c386fe39a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc1ebf3926980adf91187c154c1646c386fe39a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
