)]}'
{"id":"openstack%2Freleases~721417","triplet_id":"openstack%2Freleases~master~I891bf1ff7bc21d7994fe0f7f6779de2cc71f40ac","project":"openstack/releases","branch":"master","topic":"ussuri-rc","hashtags":[],"change_id":"I891bf1ff7bc21d7994fe0f7f6779de2cc71f40ac","subject":"Release masakari for Ussuri","status":"MERGED","created":"2020-04-21 01:59:42.000000000","updated":"2020-04-23 15:34:47.000000000","submitted":"2020-04-23 15:34:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721417-1587656087897-5004bf0a","meta_rev_id":"21820eaa81b7003da6b6702307849e3119526449","_number":721417,"virtual_id_number":721417,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 15:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"date":"2020-04-23 14:06:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-23 15:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-04-23 14:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-04-23 14:11:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-23 15:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-23 14:14:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 14:06:53.000000000","updated_by":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 14:11:18.000000000","updated_by":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"reviewer":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"state":"REVIEWER"},{"updated":"2020-04-23 14:22:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-23 15:14:02.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-23 15:34:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"386018725f7deada283758a3da2615be44db1d6f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-21 01:59:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f7bab7aa06335d8c10c239fc3362fe1730ae222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:08:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8cee3679b2e47e681a374393cb92289 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d050f08b30a749fe81e4b10007b4acab : SUCCESS in 4m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/94b43c8fbceb43668a15d1f067ab69a1 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b3c2e249e951d30c4623f8bdef0a0e060dfcb0f6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-23 12:08:55.000000000","message":"Patch Set 1: Code-Review+1\n\nPy2 support dropping LGTM to justify a major release.\nWaiting for PTL review.","accounts_in_message":[],"_revision_number":1},{"id":"4045980d81561a36683286ad21e1d66c5db2402d","author":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-23 14:06:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f60db7d3e38a93939d32e39f229e27a9175b6411","author":{"_account_id":25267,"name":"NTT system-fault-ci masakari-integration-ci","email":"masakari.integration.test@gmail.com","username":"masakari-integration-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-23 14:07:59.000000000","message":"Removed Code-Review+1 by NTT SystemFault MasakariIntegration CI \u003cmasakari.integration.test@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"2bb49cf1c88683fea50fbaf446ab5aefe775d724","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2020-04-23 14:11:18.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"7c7df096ed708049ef73912c4fb2821466f2a68f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 14:14:08.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"b1f7fce2bfd83a2da130f5231ade3ca02c4fef95","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-23 14:22:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33bfad2e3eb81ea503ed6b8870a649a4cd0d7fca","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-23 15:14:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20098ad199f401e09b895384f583481cd618ad8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:14:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f56aa2ede1105330303c1465ca174d23a4ff647","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:34:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee03ba47f094c8e905ae3fe3fcf3e33 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f013272cc77b4545b411e287fad51984 : SUCCESS in 7m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5244d3482f844530addb8b0d7770b198 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"40bee69454f24a1f1cc4395f757817cc0b91e865","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:34:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e36dac411a07afefd8eb6a59fe59b10fbb684011","revisions":{"e36dac411a07afefd8eb6a59fe59b10fbb684011":{"kind":"REWORK","_number":1,"created":"2020-04-21 01:59:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/17/721417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/721417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/721417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/721417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/721417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/721417/1"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:59:40.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 01:59:40.000000000","tz":-300},"subject":"Release masakari for Ussuri","message":"Release masakari for Ussuri\n\nRC1 release for masakari. Please +1 if things are\nready to go, or -1 if the team needs a little more\ntime. If so, please update this patch with the new\ncommit hash to use when ready.\n\nIf no response, this patch will be approved soon\nafter the April 23 deadline.\n\nChange-Id: I891bf1ff7bc21d7994fe0f7f6779de2cc71f40ac\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e36dac411a07afefd8eb6a59fe59b10fbb684011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e36dac411a07afefd8eb6a59fe59b10fbb684011"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
