)]}'
{"id":"openstack%2Freleases~721489","triplet_id":"openstack%2Freleases~master~I05ce2b4086ca8559324466c857d83d11ee7236b8","project":"openstack/releases","branch":"master","topic":"ussuri-c-a","hashtags":[],"change_id":"I05ce2b4086ca8559324466c857d83d11ee7236b8","subject":"Release manila-tempest-plugin for ussuri","status":"MERGED","created":"2020-04-21 02:24:04.000000000","updated":"2020-04-27 23:03:10.000000000","submitted":"2020-04-27 23:03:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721489-1588028590992-0ef6f594","meta_rev_id":"9b46d223705594e120d17a08856d4998bd5ec449","_number":721489,"virtual_id_number":721489,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 23:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-27 22:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-27 22:38:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-27 22:38:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-27 22:39:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 11:10:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-21 17:30:37.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-27 22:38:25.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-27 23:03:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5357354016bcfa24cf05f3d3330a162915b07a9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-21 02:24:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaecf18a7b50464e488c2c83b0a5a97c0407717f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:34:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40a33ea632f94226b778b2449a034c73 : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54b2df695cc74fb2809e720ca181180e : SUCCESS in 5m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/77f7d01d7b1f4bf592e1af5fa304e068 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"58074552933cf4141d2684c9925208f7542a27e2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-21 11:10:07.000000000","message":"Patch Set 1: Code-Review+1\n\nWaiting for PTL approval\n\nJust a warn about the bump from 0.4.1 to 1.0.0 which reflect a stable project","accounts_in_message":[],"_revision_number":1},{"id":"f36d19d7abeeed05ab845f7d770ec62596294122","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-21 17:30:37.000000000","message":"Patch Set 1: Code-Review-1\n\nWe have some changes close to merge so it would be nice to update the SHA to include them in a couple days.","accounts_in_message":[],"_revision_number":1},{"id":"86fc79ae1948157d7658f5b3f537e7c450ab5020","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-23 16:58:00.000000000","message":"Patch Set 1:\n\nHi Sean, \n\nWe\u0027re tracking a couple of feature test changes to this repository due this week; so the earliest we might have a release is next week (Apr 27-30). I was wondering if we can make a release now and follow up with a new release once we merge tests for new features?","accounts_in_message":[],"_revision_number":1},{"id":"3fd826599608f4c3a7099529672f2492e309bdf4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-23 17:06:34.000000000","message":"Patch Set 1:\n\n\u003e Hi Sean,\n \u003e \n \u003e We\u0027re tracking a couple of feature test changes to this repository\n \u003e due this week; so the earliest we might have a release is next week\n \u003e (Apr 27-30). I was wondering if we can make a release now and\n \u003e follow up with a new release once we merge tests for new features?\n\nThanks Goutham. Since this is just the tempest plugin, I think it\u0027s low risk to just hold on this one a little bit longer to give the team some more time. I would recommend trying more for the 27th than 30th though.","accounts_in_message":[],"_revision_number":1},{"id":"be538fbe71d225221876d77caa2e1712edeabac3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-23 17:39:47.000000000","message":"Patch Set 1:\n\n\u003e Thanks Goutham. Since this is just the tempest plugin, I think it\u0027s low risk to just hold on this one a little bit longer to give the team some more time. I would recommend trying more for the 27th than 30th though.\n\nThanks Sean. I\u0027ll try and get this done by 27th","accounts_in_message":[],"_revision_number":1},{"id":"131ebce597c1694c2b659df31c52949a3d8d98d3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 22:37:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9b55e1a94598fc64150615ef33305368c11794a0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 22:38:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Goutham!","accounts_in_message":[],"_revision_number":2},{"id":"a10be255ef4e7e95b3200f0ad45f691050b621c7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 22:38:25.000000000","message":"Patch Set 2: Code-Review+1\n\nThank you for your patience with this, this is now good to go!","accounts_in_message":[],"_revision_number":2},{"id":"8c9281ee3df9a4b02adf69d050de33cd5a7fb699","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:39:04.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ddbd8a4889e0863908a74d2fa27bccc15277baa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:51:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48444fd5a714443f949e180a5314903a : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a111a6274a14263b6a0f148f384f90a : SUCCESS in 5m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ac45c3ea5d2d4c459e02fb2bb84db8fe : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"af8327f2bdacb52cc31bad0c2a6daf983132b278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:54:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63b953cd0dc85e317655d22df103c14fa63ecbce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 23:03:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d96e4a88d0547aea18b557ed1bf9b79 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fc1e5b9ae54c8a9a0ebe474e70dbfa : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1006f78c0c054237b7947540558302ef : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d2c6bb605ce9f8e29f68c213ab5a0b83f691ebab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 23:03:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8709767d4058ca464badf98387127397dc7eab36","revisions":{"5633bcf6afa984bd28d5231512cef476b042e177":{"kind":"REWORK","_number":1,"created":"2020-04-21 02:24:04.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/89/721489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/721489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/721489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/721489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/721489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/721489/1"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 02:24:02.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 02:24:02.000000000","tz":-300},"subject":"Release manila-tempest-plugin for ussuri","message":"Release manila-tempest-plugin for ussuri\n\nThis is the cycle-automatic release for manila-tempest-plugin.\nPlease +1 if this is ready to go, or -1 if the team needs\na little more time. If so, please update with the new commit\nhash to use when ready.\n\nIf no response, this will be approved soon after the April\n23rd deadline.\n\nChange-Id: I05ce2b4086ca8559324466c857d83d11ee7236b8\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5633bcf6afa984bd28d5231512cef476b042e177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5633bcf6afa984bd28d5231512cef476b042e177"}]},"branch":"refs/heads/master"},"8709767d4058ca464badf98387127397dc7eab36":{"kind":"REWORK","_number":2,"created":"2020-04-27 22:37:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/721489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/721489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/721489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/721489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/721489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/721489/2"}}},"commit":{"parents":[{"commit":"634cea8ab5ac03e3e5a01baf17752334cef3f2d1","subject":"Merge \"Update docs building, Python 3 cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634cea8ab5ac03e3e5a01baf17752334cef3f2d1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-21 02:24:02.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-27 22:37:52.000000000","tz":0},"subject":"Release manila-tempest-plugin for ussuri","message":"Release manila-tempest-plugin for ussuri\n\nThis is the cycle-automatic release for manila-tempest-plugin.\n\nChange-Id: I05ce2b4086ca8559324466c857d83d11ee7236b8\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8709767d4058ca464badf98387127397dc7eab36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8709767d4058ca464badf98387127397dc7eab36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
