)]}'
{"id":"openstack%2Freleases~721883","triplet_id":"openstack%2Freleases~master~Ic882a194f4f7e5ce95dc95f81310247df65c915f","project":"openstack/releases","branch":"master","topic":"nova-train-20.2.0","hashtags":[],"change_id":"Ic882a194f4f7e5ce95dc95f81310247df65c915f","subject":"nova: Release 20.2.0","status":"MERGED","created":"2020-04-22 13:02:26.000000000","updated":"2020-04-23 09:37:49.000000000","submitted":"2020-04-23 09:37:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721883-1587634669621-354c625b","meta_rev_id":"6f89d8948e4bbb7817dd459aab58352f018f1b1b","_number":721883,"virtual_id_number":721883,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 09:37:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-22 16:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-23 09:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-04-22 14:00:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":1,"date":"2020-04-22 15:02:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-23 09:16:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-22 15:05:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-23 07:30:57.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 14:00:25.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-04-22 15:02:25.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-04-22 16:17:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-23 07:30:57.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-04-23 09:16:19.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-23 09:37:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdabb3a31b34c6ad4672f51f9e01f5e6f4df5aa9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-22 13:02:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"427e5dd43630cb451b98646a261ff67dada50854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 13:37:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5113acdd5ae84dda872a38af20170501 : SUCCESS in 8m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9e898582be443928b68568ee3d20822 : SUCCESS in 26m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6aa607d1de874b56a9bfca6d6e8cba3a : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1dbbb4295da233084cf9a57d245f7224c56eb324","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-04-22 14:00:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nLGTM, agree with the minor version bump given the API fixes and retry configurable.","accounts_in_message":[],"_revision_number":1},{"id":"7efafce3a8151ba7ce086dc228e6260ab598118b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-22 15:02:25.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks guys","accounts_in_message":[],"_revision_number":1},{"id":"89b6080b4f93f4e8062351dffd04629deb3874d6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 15:05:28.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"214a89f082d50076b8b224f7e22772be88703621","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-22 15:54:50.000000000","message":"Topic set to nova-train-20.2.0","accounts_in_message":[],"_revision_number":1},{"id":"ef950ad233187739bb45f2027e2e7f1d33c710a6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-22 16:17:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf0a135ee78e4c6e468b0127b773684792aed6ed","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-23 09:16:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee0500468dc3d834c11a256e4ff1f6fcdc27facf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:16:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f84ea773250816dcae4d86a06fb930720cc7142","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:37:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6f89d8948e4bbb7817dd459aab58352f018f1b1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:37:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e69a8dfdd864261a403a511d46a4f79 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22235a82f7134508a2304e435f159085 : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8239f6aa0cb446c6a485489d3102d729 : SUCCESS in 19m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"850216da456d3db4230fa9531dbd618d0b553c2d","revisions":{"850216da456d3db4230fa9531dbd618d0b553c2d":{"kind":"REWORK","_number":1,"created":"2020-04-22 13:02:26.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/83/721883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/721883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/721883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/721883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/721883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/721883/1"}}},"commit":{"parents":[{"commit":"4d738658b9b580717dd6442f95d83cbc43b68bcb","subject":"Merge \"Release final monasca-common for ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4d738658b9b580717dd6442f95d83cbc43b68bcb"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-04-22 11:38:58.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2020-04-22 12:45:19.000000000","tz":120},"subject":"nova: Release 20.2.0","message":"nova: Release 20.2.0\n\nMinor version bump is due to a couple of API related changes:\nIdf84ccff254d26fa13473fe9741ddac21cbcf321 / 240d030902\nI85423c7bcacff3bc465c22686d0675529d211b59 / 938b499b1f\nI0ca50665b86b9fdb4618192d4d6a3bcaa6ea2291 / 09059de8a9\na new neutron API call retry config option:\nIfb3afb13aff7e103c2e80ade817d0e63b624604a / 71971c2062\nand a live migration behavior change:\nI900f74d482fc87da5b1b5ec9db2ad5aefcfcfe7a / b08d59ac7b\n\nThe complete list of included changes:\n\n$ git log --oneline --no-merges 20.1.1..HEAD\n240d030902 Reject boot request for unsupported images\n923a4f9be6 Avoid spurious error logging in _get_compute_nodes_in_db\n6fa8540f2c Abort live-migration during instance_init\n03fd4ac755 libvirt: Calculate disk_over_committed for raw instances\n6749bad463 Use placement stable version for functional job\n88205a4e91 Reset the cell cache for database access in Service\n71971c2062 Add config option for neutron client retries\nb08d59ac7b bug-fix: Reject live migration with vpmem\n91d410b92f nova-live-migration: Ensure subnode is fenced during evacuation testing\n70447bca2f nova-live-migration: Wait for n-cpu services to come up after configuring Ceph\n588b0484bf Lowercase ironic driver hash ring and ignore case in cache\nc48d621843 Fix intermittently failing regression case\n93f903d794 Fix os-keypairs pagination links\n2f4e3ac3b0 Fix hypervisors paginted collection_name.\ne6b749dbdd Clean up allocation if unshelve fails due to neutron\nbd1bfc13d7 Reproduce bug 1862633\n3e935325a8 Unplug VIFs as part of cleanup of networks\n24600430a2 Functional test for UnexpectedDeletingTaskStateError\n73d9b6e5f6 libvirt: Ignore DiskNotFound during update_available_resource\n7fd41e9b8c libvirt: Provide the backing file format when creating qcow2 disks\n09059de8a9 Recalculate \u0027RequestSpec.numa_topology\u0027 on resize\n03d59e2893 Make RBD imagebackend flatten method idempotent\n52426cb15d Set instance CPU policy to \u0027share\u0027 through image property\n938b499b1f Ensure source service is up before resizing/migrating\n48f9b1c2ce Print help if nova-manage subcommand is not specified\n6484d9ff5b Create instance action when burying in cell0\n7d7a3ba70a libvirt: avoid cpu check at s390x arch\n\nChange-Id: Ic882a194f4f7e5ce95dc95f81310247df65c915f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/850216da456d3db4230fa9531dbd618d0b553c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/850216da456d3db4230fa9531dbd618d0b553c2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
