)]}'
{"id":"openstack%2Freleases~723560","triplet_id":"openstack%2Freleases~master~Ic2bbe41c7e66e7d81ed177ee8dc94796aba0d3e8","project":"openstack/releases","branch":"master","topic":"adddoc","hashtags":[],"change_id":"Ic2bbe41c7e66e7d81ed177ee8dc94796aba0d3e8","subject":"Add doc related to ./tools/add_release_note_links.sh","status":"MERGED","created":"2020-04-27 14:22:41.000000000","updated":"2020-04-27 16:07:27.000000000","submitted":"2020-04-27 16:07:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723560-1588003647681-2ccb4115","meta_rev_id":"15aadc502357891ec2494f8f7ec0645cec8d1477","_number":723560,"virtual_id_number":723560,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 16:07:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-27 15:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-27 15:47:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 15:47:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-27 16:07:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4c77f510fcb2b0fb4360deaeb54a3fd5f602b34","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-27 14:22:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe1c1afc230e00ec91068a91033d88af8e219b82","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 14:39:46.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for adding this. One minor typo.","accounts_in_message":[],"_revision_number":1},{"id":"b32410243434598d488a1487d1be1f093f271de1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 14:45:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5cce80f56cf47739e530e29ffe68d02 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"989de2c75f68c63b8be6f0f6be910c001acf75a3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-27 14:46:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f177f73ac6425bcb2c12e009ca09e4e6e0c2492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:03:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1aae8acac78493dbc1b3d6e1bdb1707 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c8ad798bfd529b6d3a9e93c6cb5cdcd01b8b03f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 15:47:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"92dcd79c425e4f90c3fb6290ba1483f7d8bd4260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:47:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c5a94b86cf000e6db2bbb24258198a8e854a3b38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:07:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1483bc5ed218408da09b3353bd6e414e : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413afaec596344308c899bba4c726934 : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b0b40400379bea4e84324190cf768751716639ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:07:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3853711a2c12180bfc6861822062eac580ada665","revisions":{"a0912d43a5c6b365423f607cb348f991f55d6866":{"kind":"REWORK","_number":1,"created":"2020-04-27 14:22:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/60/723560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/723560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/723560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/723560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/723560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/723560/1"}}},"commit":{"parents":[{"commit":"90010ce8bda9b45c660034c7d68d4bb607a53571","subject":"Merge \"Release senlin for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90010ce8bda9b45c660034c7d68d4bb607a53571"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-27 14:21:23.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-27 14:22:28.000000000","tz":120},"subject":"Add doc related to ./tools/add_release_note_links.sh","message":"Add doc related to ./tools/add_release_note_links.sh\n\nChange-Id: Ic2bbe41c7e66e7d81ed177ee8dc94796aba0d3e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0912d43a5c6b365423f607cb348f991f55d6866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0912d43a5c6b365423f607cb348f991f55d6866"}]},"branch":"refs/heads/master"},"3853711a2c12180bfc6861822062eac580ada665":{"kind":"REWORK","_number":2,"created":"2020-04-27 14:46:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/60/723560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/723560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/723560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/723560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/723560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/723560/2"}}},"commit":{"parents":[{"commit":"90010ce8bda9b45c660034c7d68d4bb607a53571","subject":"Merge \"Release senlin for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90010ce8bda9b45c660034c7d68d4bb607a53571"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-27 14:21:23.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-27 14:45:46.000000000","tz":120},"subject":"Add doc related to ./tools/add_release_note_links.sh","message":"Add doc related to ./tools/add_release_note_links.sh\n\nChange-Id: Ic2bbe41c7e66e7d81ed177ee8dc94796aba0d3e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3853711a2c12180bfc6861822062eac580ada665"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3853711a2c12180bfc6861822062eac580ada665"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
