)]}'
{"id":"openstack%2Freleases~723662","triplet_id":"openstack%2Freleases~master~I70bf90419419aead5233e16ff4db964e17ac68d7","project":"openstack/releases","branch":"master","topic":"release","hashtags":[],"change_id":"I70bf90419419aead5233e16ff4db964e17ac68d7","subject":"Bump vitrageversion to 7.1.0 and release ussuri","status":"MERGED","created":"2020-04-27 19:23:52.000000000","updated":"2020-04-27 22:41:25.000000000","submitted":"2020-04-27 22:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723662-1588027285007-71cafbf4","meta_rev_id":"1d48d152c8ff55c3fc3c520682fdf33b40689d96","_number":723662,"virtual_id_number":723662,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 22:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-27 19:50:16.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-27 22:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-27 22:30:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-27 19:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 19:30:08.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-27 22:30:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-27 22:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1860685256086777ed209a48c65d568bc761c949","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-27 19:23:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1d550a11e18cd75c71053440a91c7e79057af06","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:27:20.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"cb4b25f5507f3708eecf6e7f033023873fb2af7d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 19:30:08.000000000","message":"Patch Set 1:\n\ndo you want to create ussuri branches as well? You need to configure them....","accounts_in_message":[],"_revision_number":1},{"id":"1814e40ba47b8f79c93493d25310d296fef52fd5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 19:31:46.000000000","message":"Patch Set 1:\n\n\u003e do you want to create ussuri branches as well? You need to\n \u003e configure them....\n\nWas just going to comment the same thing. ;)","accounts_in_message":[],"_revision_number":1},{"id":"cbfb271538114ec0b1e99f7629c12b20b6aa77ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:35:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e5751c86434fb2a659956987e960f7 : SUCCESS in 9m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f8f6dd8d1d54c6b849340e785897030 : SUCCESS in 6m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/328b19ccaace4a6f8c9d8eda40f3aec0 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b68b7bfbc9a8a969f8d4e18c7be14798b61637cf","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-27 19:46:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eac1852586b0d6891cfae7b60112e38b406375d5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-27 19:46:40.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"46dd1d33da0c3c2133df07d63fac4c6213ea9f55","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:48:26.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"5c531ff1eb7791f39e6b15b0641ae9c581a4f066","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-27 19:50:16.000000000","message":"Patch Set 3:\n\n\u003e do you want to create ussuri branches as well? You need to\n \u003e configure them....\n\ndid it","accounts_in_message":[],"_revision_number":3},{"id":"40bbd1086205052c83aa60d636c92227d8c1e557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:56:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98694c45c3c4863a5cc43a04b919d53 : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/900a750903e1492ab9c1df4cffb29a59 : SUCCESS in 8m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0779ed04aa5d431da0f2432fccdba7d1 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"da9911d5cd5dcfa52fcfb9fb1f09aac95dd3327a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 22:30:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7a856f3c18faa989e5865d3c6e8170affa2fe989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:31:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"26a9faf8bdc1f364f6fb050c8714223ee4fdca12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:41:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fd6a5e614d43b682187e0a1b47b334 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a97751ea5994412ba985c8e19742257 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c453e139283844509ae87f0a94472231 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":3},{"id":"31844a9e1a235e7c9b7ffc6f04c919a56656d360","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:41:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"47b88726390418e20aaebebd5cac2496f4b2c4dc","revisions":{"bfa20bb440c4dc9bcfbe53576edd000c6a457ca0":{"kind":"REWORK","_number":1,"created":"2020-04-27 19:23:52.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/62/723662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/723662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/723662/1"}}},"commit":{"parents":[{"commit":"90010ce8bda9b45c660034c7d68d4bb607a53571","subject":"Merge \"Release senlin for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90010ce8bda9b45c660034c7d68d4bb607a53571"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:23:28.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:23:28.000000000","tz":180},"subject":"Bump vitrageversion to 7.1.0","message":"Bump vitrageversion to 7.1.0\n\nChange-Id: I70bf90419419aead5233e16ff4db964e17ac68d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfa20bb440c4dc9bcfbe53576edd000c6a457ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bfa20bb440c4dc9bcfbe53576edd000c6a457ca0"}]},"branch":"refs/heads/master"},"588ad133f9838b74be83d74d90b96abd39df947b":{"kind":"REWORK","_number":2,"created":"2020-04-27 19:46:12.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/62/723662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/723662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/723662/2"}}},"commit":{"parents":[{"commit":"90010ce8bda9b45c660034c7d68d4bb607a53571","subject":"Merge \"Release senlin for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90010ce8bda9b45c660034c7d68d4bb607a53571"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:23:28.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:46:09.000000000","tz":0},"subject":"Bump vitrageversion to 7.1.0","message":"Bump vitrageversion to 7.1.0\n\nChange-Id: I70bf90419419aead5233e16ff4db964e17ac68d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/588ad133f9838b74be83d74d90b96abd39df947b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/588ad133f9838b74be83d74d90b96abd39df947b"}]},"branch":"refs/heads/master"},"47b88726390418e20aaebebd5cac2496f4b2c4dc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-27 19:46:40.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/62/723662/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/62/723662/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/62/723662/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/62/723662/3"}}},"commit":{"parents":[{"commit":"90010ce8bda9b45c660034c7d68d4bb607a53571","subject":"Merge \"Release senlin for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90010ce8bda9b45c660034c7d68d4bb607a53571"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:23:28.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:46:37.000000000","tz":0},"subject":"Bump vitrageversion to 7.1.0 and release ussuri","message":"Bump vitrageversion to 7.1.0 and release ussuri\n\nChange-Id: I70bf90419419aead5233e16ff4db964e17ac68d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47b88726390418e20aaebebd5cac2496f4b2c4dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47b88726390418e20aaebebd5cac2496f4b2c4dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
