)]}'
{"id":"openstack%2Freleases~723669","triplet_id":"openstack%2Freleases~master~I0e2d9112ff4c1a751ac00d488ad207852e92052f","project":"openstack/releases","branch":"master","topic":"create_ussri_","hashtags":[],"change_id":"I0e2d9112ff4c1a751ac00d488ad207852e92052f","subject":"Release vitrage-dashboard for Ussuri","status":"MERGED","created":"2020-04-27 19:49:51.000000000","updated":"2020-04-27 22:37:30.000000000","submitted":"2020-04-27 22:37:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723669-1588027050535-92884968","meta_rev_id":"16613d38b8d01778b18e0be52366449197ec1949","_number":723669,"virtual_id_number":723669,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 22:37:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-27 22:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-04-27 22:30:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-27 20:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 22:30:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-27 22:37:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54986082de1e15acd25120b30ff900f2dcc6c048","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-27 19:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f7d04b8881384a90f4b4f77b93cc30f81e91f0d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:52:37.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2fb81388f5364e93a7a380f420557aea6ef069e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:00:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1597c72512e94324883fb309a7ac7b2d : SUCCESS in 9m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4bdc34184d2c4bb1b3151943391ecb4c : FAILURE in 3m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7ee975c37a724191b6a2902b2ba8157c : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"dc3edfad0cf0aa9a09f08f5d09f85fa3d3a1795e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-27 20:00:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f9bc0cab93ab98ed28f098d59bc021d8082acaf3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:02:39.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"149368428f04fc440263ef82f3bedb507cb86e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:17:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161d693474be449e9014480842393ce1 : SUCCESS in 8m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a7e53fbb0094d0bb7a4d45c260bb924 : SUCCESS in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d006b52b22474be4b96d9454f25b232f : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ea3495af53079b08c3d464b785bb41bd8fcc1c8e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-27 22:30:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"892f0730d297b7bc278ec3b528995ba6d8f900b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:30:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec7a7767f752374c56d863f723802c4d608ca6f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:37:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"722605d866ac28bb504ffe55ed0ac4b6a47aa81c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:37:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3badda360b64b13a0eb24658fc36351 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72ee10524ce44cda93311b7529ef29a6 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39a14c74f671453c9d418745296c4a21 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d18807d5029d6c0abbf891ec4b37fcb049e65383","revisions":{"dc50550d0751ab17eb55f255fa86d2d93b5a17b0":{"kind":"REWORK","_number":1,"created":"2020-04-27 19:49:51.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/69/723669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/723669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/723669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/723669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/723669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/723669/1"}}},"commit":{"parents":[{"commit":"90010ce8bda9b45c660034c7d68d4bb607a53571","subject":"Merge \"Release senlin for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90010ce8bda9b45c660034c7d68d4bb607a53571"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:49:34.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:49:34.000000000","tz":180},"subject":"Release vitrage-dashboard for Ussuri","message":"Release vitrage-dashboard for Ussuri\n\nChange-Id: I0e2d9112ff4c1a751ac00d488ad207852e92052f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc50550d0751ab17eb55f255fa86d2d93b5a17b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dc50550d0751ab17eb55f255fa86d2d93b5a17b0"}]},"branch":"refs/heads/master"},"d18807d5029d6c0abbf891ec4b37fcb049e65383":{"kind":"REWORK","_number":2,"created":"2020-04-27 20:00:14.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/69/723669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/723669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/723669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/723669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/723669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/723669/2"}}},"commit":{"parents":[{"commit":"90010ce8bda9b45c660034c7d68d4bb607a53571","subject":"Merge \"Release senlin for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/90010ce8bda9b45c660034c7d68d4bb607a53571"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 19:49:34.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-27 20:00:12.000000000","tz":0},"subject":"Release vitrage-dashboard for Ussuri","message":"Release vitrage-dashboard for Ussuri\n\nChange-Id: I0e2d9112ff4c1a751ac00d488ad207852e92052f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d18807d5029d6c0abbf891ec4b37fcb049e65383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d18807d5029d6c0abbf891ec4b37fcb049e65383"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
