)]}'
{"id":"openstack%2Freleases~724099","triplet_id":"openstack%2Freleases~master~I0385a3cbcc1cf771c9f3b7caef11ee46463514a6","project":"openstack/releases","branch":"master","topic":"adddoc","hashtags":[],"change_id":"I0385a3cbcc1cf771c9f3b7caef11ee46463514a6","subject":"add useful links for R-2 3rd step is doc","status":"MERGED","created":"2020-04-28 17:20:50.000000000","updated":"2020-04-30 06:52:42.000000000","submitted":"2020-04-30 06:52:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724099-1588229562395-92884968","meta_rev_id":"923c8c33e21eeffb78bd6a17297540480126c15d","_number":724099,"virtual_id_number":724099,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 06:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-29 15:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-30 06:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-30 06:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 15:25:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-30 06:38:20.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-04-30 06:52:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dc69d3b1bcf0dc418e34fb6eed071f02bebbed2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-28 17:20:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"896e83cf326e3678170f4ac4c7adee01f281a791","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-28 17:21:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e30eb52b8699c4b6c182ac4670ae49f21fd67c3d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-28 17:31:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8a57e210c35d30ba7f6cfd95172b3d00dd6c79c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:44:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf1873a5bb84a6b8ce3ca6de82e4ca1 : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":3},{"id":"1382b2da9cf200b065c3d649b201801d1c0f5492","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-29 08:36:04.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4dc50f8e3f6b12b0ff1d741f132b0178b1b86a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 08:53:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44cb36b8c6f44e0aa3b509e05ccc1d89 : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":4},{"id":"312f7a796602de4df8612b066dacf18b9333ccbd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-29 12:24:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"95e02047e885d7c80b27256055e5d0389b9dcf83","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-29 13:22:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0cf4156701e5ff14cec86fbb4c2e364ebdbc7a50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:29:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a17f9d27f846cc97501dcf8a489e08 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":5},{"id":"d94ce3d258c2c109d29246496d92c6de6da8b5ef","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-29 13:34:43.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f1eee55e93b497fa9a165c0fe7e6c76c505a73ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:43:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6251d4d75d58451999211d0fa71169d4 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":6},{"id":"186bb2327e6d720bc15ff5770ba4645daef51ead","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-29 15:25:22.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"0fab95c41ff301a057f79801d50f3669d35b2d4d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-30 06:38:20.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1718c61e8d9a8cfe1959b1d0547ca80f234fa115","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:38:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0f9b372853226b2d16365b7269e47d271b9d98c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:52:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"9dcfb2d1542857a5ade4c2bd29e21791742ec758","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 06:52:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bec0c4517254f479670efaf2c1846ad : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32b2de6f58248ffb3e09eb188749521 : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c3ba3e2bad279eba93dc0475a6e947772a2021ff","revisions":{"69b81699f7d83b8b0898239e1cadc4a53d2344c6":{"kind":"REWORK","_number":1,"created":"2020-04-28 17:20:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/724099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/724099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/724099/1"}}},"commit":{"parents":[{"commit":"7ca7d1ff11a7645a48861e537b2c3d8329898bc5","subject":"add email template related to R-2 and requirements update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ca7d1ff11a7645a48861e537b2c3d8329898bc5"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:20:35.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:20:35.000000000","tz":120},"subject":"add useful links for R-2 3rd steps is doc","message":"add useful links for R-2 3rd steps is doc\n\nChange-Id: I0385a3cbcc1cf771c9f3b7caef11ee46463514a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/69b81699f7d83b8b0898239e1cadc4a53d2344c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/69b81699f7d83b8b0898239e1cadc4a53d2344c6"}]},"branch":"refs/heads/master"},"0ad8a9d21b08aa8f5768dd23b54b6f5710f12750":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-28 17:21:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/724099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/724099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/724099/2"}}},"commit":{"parents":[{"commit":"7ca7d1ff11a7645a48861e537b2c3d8329898bc5","subject":"add email template related to R-2 and requirements update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ca7d1ff11a7645a48861e537b2c3d8329898bc5"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:20:35.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:21:31.000000000","tz":120},"subject":"add useful links for R-2 3rd step is doc","message":"add useful links for R-2 3rd step is doc\n\nChange-Id: I0385a3cbcc1cf771c9f3b7caef11ee46463514a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ad8a9d21b08aa8f5768dd23b54b6f5710f12750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0ad8a9d21b08aa8f5768dd23b54b6f5710f12750"}]},"branch":"refs/heads/master"},"ef389b88faf32e3a918ac16098cfb3529266a491":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-28 17:31:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/724099/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/724099/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/724099/3"}}},"commit":{"parents":[{"commit":"3c6bbf21eecc68fe7f8961e175c5c3b5f3d834b3","subject":"add email template related to R-2 and requirements update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c6bbf21eecc68fe7f8961e175c5c3b5f3d834b3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:20:35.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:31:24.000000000","tz":120},"subject":"add useful links for R-2 3rd step is doc","message":"add useful links for R-2 3rd step is doc\n\nChange-Id: I0385a3cbcc1cf771c9f3b7caef11ee46463514a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ef389b88faf32e3a918ac16098cfb3529266a491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ef389b88faf32e3a918ac16098cfb3529266a491"}]},"branch":"refs/heads/master"},"c6231aae9e8c90c4ac5c2c688b10b513b009be5e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-29 08:36:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/724099/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/724099/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/724099/4"}}},"commit":{"parents":[{"commit":"322ab426536c39cf4ae0a16f3e02dca33c501a79","subject":"add email template related to R-2 and requirements update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/322ab426536c39cf4ae0a16f3e02dca33c501a79"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:20:35.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-29 08:35:41.000000000","tz":120},"subject":"add useful links for R-2 3rd step is doc","message":"add useful links for R-2 3rd step is doc\n\nChange-Id: I0385a3cbcc1cf771c9f3b7caef11ee46463514a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6231aae9e8c90c4ac5c2c688b10b513b009be5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6231aae9e8c90c4ac5c2c688b10b513b009be5e"}]},"branch":"refs/heads/master"},"985f3b7fc2aa14ffd42a694eb0da098a3bd50334":{"kind":"REWORK","_number":5,"created":"2020-04-29 13:22:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/724099/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/724099/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/724099/5"}}},"commit":{"parents":[{"commit":"be46037236554e6bb776b66e88199a4654009bba","subject":"[doc][process R-2] clarifying our role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/be46037236554e6bb776b66e88199a4654009bba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:20:35.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-29 13:21:54.000000000","tz":120},"subject":"add useful links for R-2 3rd step is doc","message":"add useful links for R-2 3rd step is doc\n\nChange-Id: I0385a3cbcc1cf771c9f3b7caef11ee46463514a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/985f3b7fc2aa14ffd42a694eb0da098a3bd50334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/985f3b7fc2aa14ffd42a694eb0da098a3bd50334"}]},"branch":"refs/heads/master"},"c3ba3e2bad279eba93dc0475a6e947772a2021ff":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-04-29 13:34:43.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/724099/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/724099/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/724099/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/724099/6"}}},"commit":{"parents":[{"commit":"5f5fbb01b50bdba2b3f3fc0b6b9f8b2db1b5707c","subject":"[doc][process R-2] clarifying our role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f5fbb01b50bdba2b3f3fc0b6b9f8b2db1b5707c"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-28 17:20:35.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-04-29 13:34:25.000000000","tz":120},"subject":"add useful links for R-2 3rd step is doc","message":"add useful links for R-2 3rd step is doc\n\nChange-Id: I0385a3cbcc1cf771c9f3b7caef11ee46463514a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c3ba3e2bad279eba93dc0475a6e947772a2021ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c3ba3e2bad279eba93dc0475a6e947772a2021ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
