)]}'
{"id":"openstack%2Freleases~725846","triplet_id":"openstack%2Freleases~master~I85b5b12b1ed12bbb9e8b67e68211f6cd6e6dc658","project":"openstack/releases","branch":"master","topic":"oslo-train","hashtags":[],"change_id":"I85b5b12b1ed12bbb9e8b67e68211f6cd6e6dc658","subject":"Oslo releases for train 2020-05-06","status":"MERGED","created":"2020-05-06 12:41:39.000000000","updated":"2020-05-08 09:27:13.000000000","submitted":"2020-05-08 09:27:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725846-1588930033179-23512575","meta_rev_id":"888f146ce98f0093f9c44febd3842863e7cb24a2","_number":725846,"virtual_id_number":725846,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-05-06 16:54:15.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 09:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-05-07 17:06:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-05-08 09:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-05-08 09:18:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-05-07 17:08:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 17:06:45.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-05-08 09:18:20.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-05-08 09:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5790a47a3bb8560d4b1d538e5a1835bb09357f45","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-06 12:41:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ebc9f49c11a62521d62a2857827655a4314acb5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:42:48.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"ee099622a17df018638b062fe05f775bb9d710c6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-06 12:46:54.000000000","message":"Patch Set 1:\n\nPlease wait for Ben\u0027s validation.","accounts_in_message":[],"_revision_number":1},{"id":"fd55f95779e6f257517f3c28d390c2f0f252b30e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-06 12:47:00.000000000","message":"Removed PTL-Approved+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":1},{"id":"447c6728f8c41773381ad1e13d6ee242c6856548","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:48:23.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"52ca5d20fd830a28d322f11775239d1776de7c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:07:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b347165fc68146c3ad0d1c0dcf96737c : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/648edd4aee714b5c85883949bc960142 : SUCCESS in 24m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d35cd974a2644bf7875d6005ae731dfb : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d062d05ff0e877aad5320c0b03df00b80a4d3f36","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-06 15:00:40.000000000","message":"Removed PTL-Approved+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":1},{"id":"c9900907e7a5578e453408f2fb6fda070b7498ef","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-05-06 16:06:15.000000000","message":"Patch Set 1:\n\nThis is probably fine, but I suspect most of these do not require release. Just spot checking the first one, the only two changes are updating .gitreview and upper-constraints. Neither of those affect anything that ends up in a release.\n\nLooking further, castellan has a legitimate change to be released, but debtcollector and mox3 do not. It would be preferrable to not do essentially a noop release for the libraries that don\u0027t need them.","accounts_in_message":[],"_revision_number":1},{"id":"a8ec512a360c7d07a83f0e2105ff7c0c89101a1b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 16:07:42.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"3ff85e7b36e708776a43abf357895ec39e90ce81","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-06 16:33:43.000000000","message":"Patch Set 1:\n\n\u003e This is probably fine, but I suspect most of these do not require\n \u003e release. Just spot checking the first one, the only two changes are\n \u003e updating .gitreview and upper-constraints. Neither of those affect\n \u003e anything that ends up in a release.\n \u003e \n \u003e Looking further, castellan has a legitimate change to be released,\n \u003e but debtcollector and mox3 do not. It would be preferrable to not\n \u003e do essentially a noop release for the libraries that don\u0027t need\n \u003e them.\n\nack thanks, then  I\u0027ll adapt this patch to only release what we need.","accounts_in_message":[],"_revision_number":1},{"id":"a9bed31db61320c9d80db9873a3d2daaeb87cd43","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-06 16:53:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1b4029305e0f02aa911f364791cefbefbb4c429","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-06 16:54:15.000000000","message":"Patch Set 2:\n\n\u003e This is probably fine, but I suspect most of these do not require\n \u003e release. Just spot checking the first one, the only two changes are\n \u003e updating .gitreview and upper-constraints. Neither of those affect\n \u003e anything that ends up in a release.\n \u003e \n \u003e Looking further, castellan has a legitimate change to be released,\n \u003e but debtcollector and mox3 do not. It would be preferrable to not\n \u003e do essentially a noop release for the libraries that don\u0027t need\n \u003e them.\n\nWell in PS2 I just embed castelan.","accounts_in_message":[],"_revision_number":2},{"id":"ce95de13b2077bbf138bcbf7377441d787a1cdde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 17:07:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e27f03edd9e4bbabdcff94f146d3923 : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/511d99eef42b4a77ab67a5630c040487 : SUCCESS in 6m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/61567aae64324cada5a731f5fb0868bc : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"fc129f6d8e29188d40ab40d05de597e0078f0417","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-05-07 17:06:45.000000000","message":"Patch Set 2: Code-Review+1\n\nCould just title the commit castellan 1.3.2, but this looks fine.","accounts_in_message":[],"_revision_number":2},{"id":"45a186f3407bd245900d4ed26319f81d52210dc8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:08:51.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"520ebb7c13cafcd9d622819b349f245068b561a6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-08 09:18:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c125f70fc4e8f5424fc8066c6c50700ce1025657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:18:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b07e5f47d71129148718e2ee6979c8e11052cda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:27:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c35488b604485eb8923f6e04807c16 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0635fd7cd84be0951a072ee0361ced : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27132ed552fd453098f00ef3c9fe6556 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"95088909f7af76a8a404a8e290666b49d4d27667","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:27:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"569a4e2ef1dac0c4b1c10f482ff94238a1ab49fc","revisions":{"a4e6e9bcab3f3925fb044653c531dd25f8643ad7":{"kind":"REWORK","_number":1,"created":"2020-05-06 12:41:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/725846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/725846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/725846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/725846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/725846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/725846/1"}}},"commit":{"parents":[{"commit":"44595097c653ad4549cda9651b81944e3a513589","subject":"Merge \"Release nova RC2 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44595097c653ad4549cda9651b81944e3a513589"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-06 12:39:57.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-06 12:39:57.000000000","tz":120},"subject":"Oslo releases for train 2020-05-06","message":"Oslo releases for train 2020-05-06\n\nChange-Id: I85b5b12b1ed12bbb9e8b67e68211f6cd6e6dc658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4e6e9bcab3f3925fb044653c531dd25f8643ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4e6e9bcab3f3925fb044653c531dd25f8643ad7"}]},"branch":"refs/heads/master"},"569a4e2ef1dac0c4b1c10f482ff94238a1ab49fc":{"kind":"REWORK","_number":2,"created":"2020-05-06 16:53:12.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/725846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/725846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/725846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/725846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/725846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/725846/2"}}},"commit":{"parents":[{"commit":"44595097c653ad4549cda9651b81944e3a513589","subject":"Merge \"Release nova RC2 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/44595097c653ad4549cda9651b81944e3a513589"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-06 16:52:13.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-06 16:52:13.000000000","tz":120},"subject":"Oslo releases for train 2020-05-06","message":"Oslo releases for train 2020-05-06\n\nChange-Id: I85b5b12b1ed12bbb9e8b67e68211f6cd6e6dc658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/569a4e2ef1dac0c4b1c10f482ff94238a1ab49fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/569a4e2ef1dac0c4b1c10f482ff94238a1ab49fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
