)]}'
{"id":"openstack%2Freleases~726129","triplet_id":"openstack%2Freleases~master~I80742ab8204fac9b7ad105a22626be4e69ff38ea","project":"openstack/releases","branch":"master","topic":"r1-final-rc-deadline","hashtags":[],"change_id":"I80742ab8204fac9b7ad105a22626be4e69ff38ea","subject":"Final RC patch for keystone","status":"MERGED","created":"2020-05-07 13:57:38.000000000","updated":"2020-05-08 09:24:05.000000000","submitted":"2020-05-08 09:24:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726129-1588929845144-95dab07b","meta_rev_id":"98dee5240a42b223f1a07435d994bc3327706cae","_number":726129,"virtual_id_number":726129,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 09:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-07 21:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-08 03:22:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2020-05-08 09:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-07 20:37:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2020-05-08 09:12:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-05-08 03:23:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 20:37:53.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-05-07 21:37:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-08 03:22:50.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-05-08 09:12:12.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-05-08 09:24:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca76218e9aa168dbf1e7e6bcdc88c3feb91d6a6a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 13:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb1fec6420f40042f3fe4d8ed99ede715db9a6e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:09:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b073a82b1994331b52ea3e29f9f9920 : SUCCESS in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1229b5ca5be547ce810604c8d4b34c3d : SUCCESS in 10m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/33b8fd3d1d9c4536a7b57b7993fd36da : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"34b3964d66590ddb14ee2e08440a91e991acdb1b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 14:12:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2b00cc9deeb1751cc36b3013d5ae8423d1afe725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:09:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e744c2dec5ff4f0698e939984d742e0e : SUCCESS in 9m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7f6593eefa2400ab4c5fb06aa91723f : SUCCESS in 9m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bad5bd67c02a4c62929824a3c497e08d : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6290c38e4f265b262974da1e9d8e0b6a95a0c574","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-07 16:10:32.000000000","message":"Patch Set 2: Code-Review-1\n\nCan we please hold this for https://review.opendev.org/725888 (OSSA-2020-004) and https://review.opendev.org/725887 (OSSA-2020-005)?","accounts_in_message":[],"_revision_number":2},{"id":"da69af9e1dea8d4ec03d295477047b03c209182d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 16:38:10.000000000","message":"Patch Set 2:\n\n\u003e Can we please hold this for https://review.opendev.org/725888\n \u003e (OSSA-2020-004) and https://review.opendev.org/725887\n \u003e (OSSA-2020-005)?\n\nSince they are security fix and that they are +2 +w 3 hours ago I think yes it look reasonable to wait for these ones.","accounts_in_message":[],"_revision_number":2},{"id":"6a046971d02e233465aa4282e89689cec4f18f03","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 20:30:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce28300cb4d3f66e99441b1acae36447d7927314","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-07 20:37:53.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks! Thumbs up from the VMT. This includes the critical security fixes we were hoping to release with.","accounts_in_message":[],"_revision_number":3},{"id":"3ac74c4d8389db1cbd9591f839e244df929cecaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 20:41:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a190af43e9494e7491cf0c691dcc0f86 : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f46a6068ab84501bc9080b71560ffbe : SUCCESS in 9m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bacd25583e3744ba9511cbb04679cb4c : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"6485d51debb8e9a5ef90657ff74e5e31a8e58d03","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-07 21:37:57.000000000","message":"Patch Set 3: Code-Review+2\n\nDoesn\u0027t look like any pending patches about to merge and identified security patches are now included. Will wait a bit for PTL approval, but I think we are good to go.","accounts_in_message":[],"_revision_number":3},{"id":"c2d1fb235046a4513d584c51980b7a032c27c6dc","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-05-08 03:22:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8bab3520c572a29fc6353d8f03f25fdd90cf5073","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 03:23:48.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"b7535ff957ed79c91f172a11d4c8c27f092f369b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-08 09:12:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a0d80b744df8b205e8b1c0d6b1f70f0af79b8363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:12:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4c5c2436ab4ab4d94d4c7b4fbd55f92ecb7c3f4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:24:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69520f57b3a541f1a7b445bff89893e9 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bb43d1c573493fafe4080a8d048226 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90804bae6ccc4c5ea1731f13b20cae85 : SUCCESS in 9m 15s","accounts_in_message":[],"_revision_number":3},{"id":"342aea401792086c0d6f7e1e988505f6e44b0557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:24:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fb4092f429f3e5ba660c28fafe84b6c2b98f5ebd","revisions":{"f0e7386d2862a74b859412bde611144ac1f850a6":{"kind":"REWORK","_number":1,"created":"2020-05-07 13:57:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/29/726129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/726129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/726129/1"}}},"commit":{"parents":[{"commit":"34add1887806cf47764288703fa9c1e7747c23b2","subject":"Final RC patch for heat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34add1887806cf47764288703fa9c1e7747c23b2"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:54:14.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:54:14.000000000","tz":120},"subject":"Final RC patch for keystone","message":"Final RC patch for keystone\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I80742ab8204fac9b7ad105a22626be4e69ff38ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0e7386d2862a74b859412bde611144ac1f850a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0e7386d2862a74b859412bde611144ac1f850a6"}]},"branch":"refs/heads/master"},"5e2abf8e5553558cc31f36b561a71bfcab7d74e1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-07 14:12:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/29/726129/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/726129/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/726129/2"}}},"commit":{"parents":[{"commit":"d6a8471de0687dc2cfb0614eb93902dfdd0ef94a","subject":"Merge \"Display more help to the new-release wrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6a8471de0687dc2cfb0614eb93902dfdd0ef94a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:54:14.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 14:12:22.000000000","tz":120},"subject":"Final RC patch for keystone","message":"Final RC patch for keystone\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I80742ab8204fac9b7ad105a22626be4e69ff38ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e2abf8e5553558cc31f36b561a71bfcab7d74e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e2abf8e5553558cc31f36b561a71bfcab7d74e1"}]},"branch":"refs/heads/master"},"fb4092f429f3e5ba660c28fafe84b6c2b98f5ebd":{"kind":"REWORK","_number":3,"created":"2020-05-07 20:30:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/29/726129/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/726129/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/726129/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/726129/3"}}},"commit":{"parents":[{"commit":"d6a8471de0687dc2cfb0614eb93902dfdd0ef94a","subject":"Merge \"Display more help to the new-release wrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6a8471de0687dc2cfb0614eb93902dfdd0ef94a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:54:14.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 20:28:54.000000000","tz":120},"subject":"Final RC patch for keystone","message":"Final RC patch for keystone\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I80742ab8204fac9b7ad105a22626be4e69ff38ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb4092f429f3e5ba660c28fafe84b6c2b98f5ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb4092f429f3e5ba660c28fafe84b6c2b98f5ebd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
