)]}'
{"id":"openstack%2Freleases~726133","triplet_id":"openstack%2Freleases~master~I2afb93e0c200450333784533855e75b2dee32173","project":"openstack/releases","branch":"master","topic":"r1-final-rc-deadline","hashtags":[],"change_id":"I2afb93e0c200450333784533855e75b2dee32173","subject":"Final RC patch for monasca-agent","status":"ABANDONED","created":"2020-05-07 13:57:38.000000000","updated":"2020-05-07 15:39:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34931b1a982f9290706881d7de3c78aa23ee226f","_number":726133,"virtual_id_number":726133,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-07 14:44:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-07 14:21:03.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:21:03.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-05-07 14:44:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"350fdb4814adbe8d008419e689eba4fbc454008a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 13:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0285fc82fa409a853f0b902a9eeb7fe573a7067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:08:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c063872198344bc0b5c706d182cbd8d1 : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60b28c88784c4d2384275f4a29c6020f : SUCCESS in 7m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d7b346b244dc41fdbd29540015ec0e49 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"33b9af60821b6392ff1f210c790ecd33a43100f3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 14:11:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7255ad4569b21a7acf7c8ae67c660b11b41ec4c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:44:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20da27d05754bb08ed1a0acbc4386e5 : SUCCESS in 9m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c53e579d0cb45e98027751551c9acd2 : SUCCESS in 6m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/62849143b52047fd8028441f28c34857 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"8229eca885144db27c57a9ccb5d2e6205680c6d5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-07 15:39:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8943be8fed845ce649ee31172c20c646ecc7ee67","revisions":{"0caf358f6b53739a2d10c758ffca8670c2504cc3":{"kind":"REWORK","_number":1,"created":"2020-05-07 13:57:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/726133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/726133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/726133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/726133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/726133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/726133/1"}}},"commit":{"parents":[{"commit":"e430c732e4484b92b9a63b87b58c3986835a1116","subject":"Final RC patch for manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e430c732e4484b92b9a63b87b58c3986835a1116"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:54:52.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:54:52.000000000","tz":120},"subject":"Final RC patch for monasca-agent","message":"Final RC patch for monasca-agent\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I2afb93e0c200450333784533855e75b2dee32173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0caf358f6b53739a2d10c758ffca8670c2504cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0caf358f6b53739a2d10c758ffca8670c2504cc3"}]},"branch":"refs/heads/master"},"8943be8fed845ce649ee31172c20c646ecc7ee67":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-07 14:11:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/726133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/726133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/726133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/726133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/726133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/726133/2"}}},"commit":{"parents":[{"commit":"d6a8471de0687dc2cfb0614eb93902dfdd0ef94a","subject":"Merge \"Display more help to the new-release wrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6a8471de0687dc2cfb0614eb93902dfdd0ef94a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:54:52.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 14:10:55.000000000","tz":120},"subject":"Final RC patch for monasca-agent","message":"Final RC patch for monasca-agent\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I2afb93e0c200450333784533855e75b2dee32173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8943be8fed845ce649ee31172c20c646ecc7ee67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8943be8fed845ce649ee31172c20c646ecc7ee67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
