)]}'
{"id":"openstack%2Freleases~726142","triplet_id":"openstack%2Freleases~master~I5fb84f8117d394ad040fb78b6d8cc78681abdb3f","project":"openstack/releases","branch":"master","topic":"r1-final-rc-deadline","hashtags":[],"change_id":"I5fb84f8117d394ad040fb78b6d8cc78681abdb3f","subject":"Final RC patch for sahara-dashboard","status":"MERGED","created":"2020-05-07 13:57:38.000000000","updated":"2020-05-07 16:00:00.000000000","submitted":"2020-05-07 16:00:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726142-1588867200256-95dab07b","meta_rev_id":"7fc9fa15b1d7988e405db86afcd683bbf62a538e","_number":726142,"virtual_id_number":726142,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 15:59:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-07 15:42:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-07 14:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-07 15:42:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"date":"2020-05-07 14:39:57.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-05-07 14:41:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:39:53.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2020-05-07 14:39:57.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-07 15:42:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-07 15:59:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"320fd2aee3d839eea7dd60c23c782a2231eb984d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 13:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66c1222d0fe5888bec4f10023832f847b312116a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 14:08:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5ee6bfc5baef716f1170861d9b7f099c5e6b2fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:19:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed33191ed09542a1a367f91dab852685 : SUCCESS in 9m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74afd52e03ea49fa812ffcbe6a53793a : SUCCESS in 8m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7c5934ffaf7d4a2199d3544d8935b823 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1e15a5bfc393ade4731c7907dc82909c7c169567","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-05-07 14:39:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b35e4b950101efe18c8b871b3594cdf9b9cea13d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:41:49.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"16b4e83eb8f75f3f59db8aa11188aa267b2a9ad7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-07 15:42:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"47bc6922ec13fef4e64a9bada18543891d54002d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:42:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"59e2b43347fc2bed5e17f1cb397a235008474849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:59:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df83003339744d6991cd5aeaff339c34 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdf4eb376d547359bc875473a810f6e : SUCCESS in 10m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80800609f164479fb34ee128fdb5fc52 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"108ae337a405903caf5e07f2f6fa11d52af55f08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 16:00:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bba08b955e9db82d6852d19ffa00db4982f183d2","revisions":{"e6272a632c9129cfea0198b10273a24385ab0c40":{"kind":"REWORK","_number":1,"created":"2020-05-07 13:57:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/42/726142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/726142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/726142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/726142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/726142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/726142/1"}}},"commit":{"parents":[{"commit":"d3b25c3471a75799e65a3cb9969864440fc06ef2","subject":"Final RC patch for octavia-dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3b25c3471a75799e65a3cb9969864440fc06ef2"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:13.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:13.000000000","tz":120},"subject":"Final RC patch for sahara-dashboard","message":"Final RC patch for sahara-dashboard\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I5fb84f8117d394ad040fb78b6d8cc78681abdb3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6272a632c9129cfea0198b10273a24385ab0c40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6272a632c9129cfea0198b10273a24385ab0c40"}]},"branch":"refs/heads/master"},"bba08b955e9db82d6852d19ffa00db4982f183d2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-07 14:08:07.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/42/726142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/726142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/726142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/726142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/726142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/726142/2"}}},"commit":{"parents":[{"commit":"d6a8471de0687dc2cfb0614eb93902dfdd0ef94a","subject":"Merge \"Display more help to the new-release wrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6a8471de0687dc2cfb0614eb93902dfdd0ef94a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:13.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 14:07:59.000000000","tz":120},"subject":"Final RC patch for sahara-dashboard","message":"Final RC patch for sahara-dashboard\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I5fb84f8117d394ad040fb78b6d8cc78681abdb3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bba08b955e9db82d6852d19ffa00db4982f183d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bba08b955e9db82d6852d19ffa00db4982f183d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
