)]}'
{"id":"openstack%2Freleases~726145","triplet_id":"openstack%2Freleases~master~I3e3d9b698cbd608d6c1f5bc896f6e02c4d8bb44a","project":"openstack/releases","branch":"master","topic":"r1-final-rc-deadline","hashtags":[],"change_id":"I3e3d9b698cbd608d6c1f5bc896f6e02c4d8bb44a","subject":"Final RC patch for searchlight","status":"MERGED","created":"2020-05-07 13:57:38.000000000","updated":"2020-05-07 15:59:10.000000000","submitted":"2020-05-07 15:59:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726145-1588867150818-2682b9c6","meta_rev_id":"f372c050fd8b83f18fba15bc6a510b2573615c79","_number":726145,"virtual_id_number":726145,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 15:59:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-07 15:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-07 14:27:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-07 15:41:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-05-07 14:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:27:32.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-05-07 15:41:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-07 15:59:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b869878344ca61c1b3c462a39a83ceca381d6187","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 13:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74845257a0a3e47a2ac9fe64069902514ef885f7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 14:07:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"72e357b2a8549bc8e4b265be44bbb1b20d5c93ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:18:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/315e235f51084ccba315913d1df3ebb6 : SUCCESS in 9m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9aae5f9e7f96469caf8e78d147d25c84 : SUCCESS in 8m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/eee137c5f0754d2b870937a18a86f58d : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":2},{"id":"3da96016826e76f96947654241354394d618112e","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-05-07 14:27:32.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":2},{"id":"aafcb49f06982717ed153edee3f0b1c7613a800a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:29:01.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3f5799f58767559a29b745af3bed3ccd4f8218b7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-07 15:41:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"93b5e1aaecbec0812a3d37f2bfa19d33528db230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:41:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5929cdae42a81c9a2d28bcc8061aa893280dbe77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:59:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"105380634f72437ae59b5e9e5bca2d92dd346ea4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:59:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da6750b71c741d2a4cd725e021edab3 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/683a467e266b4fdfb4b545d9fdbb8b4b : SUCCESS in 10m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9c57419396e435ba32f8c317e7f049e : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e06d50729db45b263f5fef97031dbaab2e5febc1","revisions":{"4ecf4bb9851df4b5b5c4704a8097893ac3778e87":{"kind":"REWORK","_number":1,"created":"2020-05-07 13:57:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/45/726145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/726145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/726145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/726145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/726145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/726145/1"}}},"commit":{"parents":[{"commit":"71008ea78bc6371d9e699aab224ab30904f03fac","subject":"Final RC patch for searchlight-ui","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71008ea78bc6371d9e699aab224ab30904f03fac"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:34.000000000","tz":120},"subject":"Final RC patch for searchlight","message":"Final RC patch for searchlight\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I3e3d9b698cbd608d6c1f5bc896f6e02c4d8bb44a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ecf4bb9851df4b5b5c4704a8097893ac3778e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ecf4bb9851df4b5b5c4704a8097893ac3778e87"}]},"branch":"refs/heads/master"},"e06d50729db45b263f5fef97031dbaab2e5febc1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-07 14:07:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/45/726145/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/726145/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/726145/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/726145/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/726145/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/726145/2"}}},"commit":{"parents":[{"commit":"d6a8471de0687dc2cfb0614eb93902dfdd0ef94a","subject":"Merge \"Display more help to the new-release wrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6a8471de0687dc2cfb0614eb93902dfdd0ef94a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:34.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 14:06:50.000000000","tz":120},"subject":"Final RC patch for searchlight","message":"Final RC patch for searchlight\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I3e3d9b698cbd608d6c1f5bc896f6e02c4d8bb44a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e06d50729db45b263f5fef97031dbaab2e5febc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e06d50729db45b263f5fef97031dbaab2e5febc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
