)]}'
{"id":"openstack%2Freleases~726147","triplet_id":"openstack%2Freleases~master~I51ae266207d5eb38b75235d685258ea111b97008","project":"openstack/releases","branch":"master","topic":"r1-final-rc-deadline","hashtags":[],"change_id":"I51ae266207d5eb38b75235d685258ea111b97008","subject":"Final RC patch for senlin","status":"MERGED","created":"2020-05-07 13:57:38.000000000","updated":"2020-05-07 17:11:45.000000000","submitted":"2020-05-07 17:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726147-1588871505970-ae45bbcb","meta_rev_id":"18eb45985843b03ad24274f5d1a2697ef37ef0bb","_number":726147,"virtual_id_number":726147,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-05-07 16:20:13.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 17:11:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-07 17:00:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"value":1,"date":"2020-05-07 16:52:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-07 17:00:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-05-07 14:21:46.000000000","_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-05-07 16:54:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 14:21:46.000000000","updated_by":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"reviewer":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"state":"REVIEWER"},{"updated":"2020-05-07 16:20:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-05-07 16:52:05.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2020-05-07 17:00:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-07 17:11:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3451b38c05b90921bbb5d7943723e74f11e280db","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 13:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"369fdea39bd9dcef86775ba2d60ee99b035c90f3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-07 14:06:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"742a99447bf965dc2f917eb5cafbc1c7d072eebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 14:23:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bdf0d1fd9384027afed73e52f68bb95 : SUCCESS in 9m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5fde9f61acc742fd8ae661b1fb75dd5c : SUCCESS in 9m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/29a54ae6c1b64fce86b469873360d6a1 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0a608756af04936596ebc94ef133bd3403a6308e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-07 16:20:13.000000000","message":"Patch Set 2:\n\nLooks like it\u0027s only translations, so probably OK to skip if we can\u0027t get PTL+1 in time.","accounts_in_message":[],"_revision_number":2},{"id":"0a175587c8b40af9b0aa40f60429829cf0d8c8cf","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2020-05-07 16:52:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"264bd42ffe0c388409e4077c0615cd59911b093c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 16:54:20.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9e05286aa603f6e06c949f9e2c63d5e90907e33d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-07 17:00:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf24ec00a7c8b13ae96f35899e1325c225fee11f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:00:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ea51df4d2288b8bec6d21b142ba1941e61673385","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:11:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f8aab14f6af9003375286385c3ab96313454bc7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:11:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50fa8bdb036344a481588ea4dfd202f3 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6e2e70befc4f7c839c3b928834ba9b : SUCCESS in 7m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e6433b268d3429a9927918ee95b1d63 : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"00a7dafc99647e57c242e4bfd901a649303087ba","revisions":{"976b13aa83411c8666e191ed106ca7e829401d0a":{"kind":"REWORK","_number":1,"created":"2020-05-07 13:57:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/726147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/726147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/726147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/726147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/726147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/726147/1"}}},"commit":{"parents":[{"commit":"0a7dcb13a7bbcf02abb0f525d7cd4607be065d68","subject":"Final RC patch for senlin-dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0a7dcb13a7bbcf02abb0f525d7cd4607be065d68"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:46.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:46.000000000","tz":120},"subject":"Final RC patch for senlin","message":"Final RC patch for senlin\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I51ae266207d5eb38b75235d685258ea111b97008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/976b13aa83411c8666e191ed106ca7e829401d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/976b13aa83411c8666e191ed106ca7e829401d0a"}]},"branch":"refs/heads/master"},"00a7dafc99647e57c242e4bfd901a649303087ba":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-07 14:06:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/726147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/726147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/726147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/726147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/726147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/726147/2"}}},"commit":{"parents":[{"commit":"d6a8471de0687dc2cfb0614eb93902dfdd0ef94a","subject":"Merge \"Display more help to the new-release wrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6a8471de0687dc2cfb0614eb93902dfdd0ef94a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 13:56:46.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-07 14:06:32.000000000","tz":120},"subject":"Final RC patch for senlin","message":"Final RC patch for senlin\n\nThis creates a new RC before final release candidate ending to release latest fixes and translations.\n\nPlease ack with +1, or if there is some reason to hold off leave a -1.\n\nChange-Id: I51ae266207d5eb38b75235d685258ea111b97008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00a7dafc99647e57c242e4bfd901a649303087ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00a7dafc99647e57c242e4bfd901a649303087ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
